)]}'
{"id":"openstack%2Fironic-lib~933511","triplet_id":"openstack%2Fironic-lib~master~Ie8b88552dbf9e1ef6fc633fe2a00b66a9c00369a","project":"openstack/ironic-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-05 12:59:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie8b88552dbf9e1ef6fc633fe2a00b66a9c00369a","subject":"Add note about requirements lower bounds","status":"MERGED","created":"2024-10-26 15:36:23.000000000","updated":"2024-11-05 13:00:26.000000000","submitted":"2024-11-05 12:59:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933511","meta_rev_id":"566d2b9d2a8a59968f3c85d8e9f35198e2dbff0e","_number":933511,"virtual_id_number":933511,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-05 12:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2024-10-26 21:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-05 12:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-05 12:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-26 17:31:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-26 21:56:20.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2024-11-05 12:10:02.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"f7e0df429122a791a8cc2c04942da5c4107ab915","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 15:36:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8a406fc2fe10c967ee6e8338e9864e04bf1d3b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 16:13:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f4e2ca7ab71dec1da0d1d2f4a8881082b7f761da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 17:31:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdcca6753b5840988ccc5f6ff3865048\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ce9fee10001c48f2bcbf55521494fad9 : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab7d6d49b0cd4239a5e6fac31ed372f7 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a9154c8376e45e785cb1d99b21c0144 : SUCCESS in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/457331bd456c48d6bfb834ee389bbd7e : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01bcfb2fcda7416eab2e1e21c6e4a981 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a931cd14f6641c6ab73dae0b06d2de1 : SUCCESS in 4m 27s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/6fe8b73a0d124329a37d6e150f313198 : SUCCESS in 39m 49s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/f9c1deddf86b4b1c9eae69f3de1c398c : SUCCESS in 1h 12m 20s\n- ironic-lib-tox-codespell https://zuul.opendev.org/t/openstack/build/db4f5b9f3fc74cb99e8b9e9ff7e8a99f : SUCCESS in 2m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50014f96b30f6ed155c2c59ce9d72cf3c08e3fcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 18:29:25.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b6d6da5587234e716c115cc962730b884fd76c4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 19:30:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee396093e25240cf92957e16d7702c18\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d9093dc399894563b176966a6fdbedb8 : SUCCESS in 1m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3edd15be0b16450fa33549cbfed45ab3 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d41d170fdd294e22a2968e45b9fbccce : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fe14aca90a745229678eadceb6c44ca : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40f0249e53a742b39905f31d30b3eeb9 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b38898ed5a4d6b9a9b48a0740a63b8 : SUCCESS in 3m 27s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/a36db0063f9e42cc8059f895e2b8fb6b : SUCCESS in 44m 08s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/72fd1266ae75458ca4199f61657bcf6f : SUCCESS in 50m 20s\n- ironic-lib-tox-codespell https://zuul.opendev.org/t/openstack/build/932cb2851dfb424f9bab053d43f2c7a8 : SUCCESS in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"20cd23163a460c8352f8fa18d325d0273710c950","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-26 21:56:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"14f483fd607b58e9eaccf39c2cc52b2713e6949b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-11-05 12:10:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"06f162cf4694b34d4f09608c8f7ecb439c3b4ef1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 12:10:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"37d7630c888b11b2fb77bc7a15e52d6ae5aa4bd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 12:59:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e490d90b6ad44c4810739c01d34d0ad\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca249f0ae7e14b76856ae9a2f1952dfe : SUCCESS in 1m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22765050ceac41c0a591548ebfd950c6 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d66d09a5f644648b985df32f7787b42 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdf3264d06084a17a4545f7ab1467519 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ddaa977b70246deab56a3d03ea5d37f : SUCCESS in 4m 34s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/767e3e3d8fb94939888018fd1d938cab : SUCCESS in 48m 22s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/c2f15fb89dc142e982574358c5462dd5 : SUCCESS in 47m 43s","accounts_in_message":[],"_revision_number":3},{"id":"5251cc7aeb3906677acb181877ae794d9c4dbdb0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 12:59:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"566d2b9d2a8a59968f3c85d8e9f35198e2dbff0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 13:00:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/035e68f2298d4610b05fdfe20aaed72e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3624b1f4125a43b783c6963f76458c33 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a3eabea14bf34cd1d48e431fae3c85d5dba0a9f2","revisions":{"0e2b2f398306df9f3a95021df5e21735029287d3":{"kind":"REWORK","_number":1,"created":"2024-10-26 15:36:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/933511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/11/933511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/1"}}},"commit":{"parents":[{"commit":"ae1368a886c20dcb61c5e8eea40ba839622a6526","subject":"add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/ae1368a886c20dcb61c5e8eea40ba839622a6526"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 15:36:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 15:36:15.000000000","tz":540},"subject":"Add note about requirements lower bounds","message":"Add note about requirements lower bounds\n\nSince lower constraints job was removed the bounds are not actually\ntested. Add a note to explain the bounds are maintained on bast effort\nbasis.\n\nChange-Id: Ie8b88552dbf9e1ef6fc633fe2a00b66a9c00369a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/0e2b2f398306df9f3a95021df5e21735029287d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/0e2b2f398306df9f3a95021df5e21735029287d3"}]},"branch":"refs/heads/master"},"978a14831025e01cbdfdd69ca46db2c3b89a228e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-26 16:13:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/933511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/11/933511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/2"}}},"commit":{"parents":[{"commit":"846baa857cb57850833a6bac52aae2f55e5782fb","subject":"Replace outdated python job template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/846baa857cb57850833a6bac52aae2f55e5782fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 15:36:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 16:13:02.000000000","tz":0},"subject":"Add note about requirements lower bounds","message":"Add note about requirements lower bounds\n\nSince lower constraints job was removed the bounds are not actually\ntested. Add a note to explain the bounds are maintained on bast effort\nbasis.\n\nChange-Id: Ie8b88552dbf9e1ef6fc633fe2a00b66a9c00369a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/978a14831025e01cbdfdd69ca46db2c3b89a228e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/978a14831025e01cbdfdd69ca46db2c3b89a228e"}]},"branch":"refs/heads/master","description":"Rebase"},"a3eabea14bf34cd1d48e431fae3c85d5dba0a9f2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-10-26 18:29:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/933511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/11/933511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/11/933511/3"}}},"commit":{"parents":[{"commit":"846baa857cb57850833a6bac52aae2f55e5782fb","subject":"Replace outdated python job template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/846baa857cb57850833a6bac52aae2f55e5782fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 15:36:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 18:29:25.000000000","tz":0},"subject":"Add note about requirements lower bounds","message":"Add note about requirements lower bounds\n\nSince lower constraints job was removed the bounds are not actually\ntested. Add a note to explain the bounds are maintained on best effort\nbasis.\n\nChange-Id: Ie8b88552dbf9e1ef6fc633fe2a00b66a9c00369a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/a3eabea14bf34cd1d48e431fae3c85d5dba0a9f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/a3eabea14bf34cd1d48e431fae3c85d5dba0a9f2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
