)]}'
{"id":"openstack%2Fironic-lib~933898","triplet_id":"openstack%2Fironic-lib~stable%2F2024.2~I7385739544ee72de3ffafb4b32611e783f4890b2","project":"openstack/ironic-lib","branch":"stable/2024.2","topic":"ironic-fix-python-job-stable/2024.2","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-09 13:09:39.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I7385739544ee72de3ffafb4b32611e783f4890b2","subject":"Replace outdated python job template","status":"MERGED","created":"2024-11-01 00:15:04.000000000","updated":"2025-02-09 13:11:23.000000000","submitted":"2025-02-09 13:09:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933898-ironic-fix-python-job-stable/2024.2","cherry_pick_of_change":933242,"cherry_pick_of_patch_set":1,"meta_rev_id":"e29f01c16fe9ef7097a8a1d3f9eda61034fbfa35","_number":933898,"virtual_id_number":933898,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-09 13:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2024-11-01 00:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2025-02-08 08:01:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":2,"date":"2025-02-08 20:44:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":1,"date":"2025-02-08 20:44:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-01 00:25:42.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2024-11-01 00:45:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-08 08:01:22.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"},{"updated":"2025-02-08 20:44:02.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"6a2951eb920f881268fc23966b894ffd76b368ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-01 00:15:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e9c239ea1179cbaf985a951d0e9364c424eceb8b","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-11-01 00:25:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"efc5e17c3812c732db1c461232fcda4c3bf63e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 00:45:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d70df15e92fc46119d3bf49ae1944ae4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87c1c1010df049a395e2ba696555f706 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba2c9ff23a14040924db0d4282ddad6 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a67aa197c5124ea99a680cf5c66680dd : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54b14bb69ecb4085baaf1fa545169dc8 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8145820a2d8e4e62a48e941d5c233acd : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151b89c94e3a4e5eaea037f5d0b05590 : SUCCESS in 5m 52s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/a72cd30e74654d219586b83665614563 : FAILURE in 27m 41s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/26899f24a3af4693b922c73db715a427 : FAILURE in 28m 58s\n- ironic-lib-tox-codespell https://zuul.opendev.org/t/openstack/build/1137b79c1d3e4d068795fe39d447b093 : SUCCESS in 2m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f8132e3984febb738b313eb6f8ad273b456e7fc","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-01-31 00:10:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"275e1d0a9454b2d5eb72d6e104cad3ef7d30885b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 01:34:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5b8c9a463d84fbab5026c34c6414433\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0c1ac6302c94c46a476167d4b12c232 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0751fd614982497e9f820f0a408a2553 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86d27dd1ae0d4921b1a03e404a2b0032 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e9f08050a9b40cfa2cd034da817767f : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84595ade722a4d9fa7b95155d74993e4 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8894f7e5c4c452b91c2d8a4fce883ad : SUCCESS in 5m 42s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/79963bf0250e427f8d64110934c4ed3f : SUCCESS in 46m 53s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/fcff6bd609d444e4a267d6b7c5f3a27a : SUCCESS in 1h 18m 16s\n- ironic-lib-tox-codespell https://zuul.opendev.org/t/openstack/build/801eb487d00e4ebb93bc44cb695bfd7e : SUCCESS in 2m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d6498e5fb0f2c92ad2f5ad69b1643542ddcd20f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-01-31 02:09:19.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"035e1fc533b3cb345c67c240ccdbc20501cf8145","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-02-08 08:01:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"171c6c90c84fb627fb9b1cd27a97219959417555","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-02-08 20:44:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22cf1f508031fe94728586bf9baafc542c98f74c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 20:44:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b9539f44dd89be3a19b95de5c3f2a050b92cab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 22:12:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ceca114f31e84e95b5d30995369fe997\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b51482a8cbc147b1b2e07d9a029fec75 : SUCCESS in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eaa2b13c62404161bad7674b011830d6 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3bf5aac3202043549258d7234b7436ae : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c1dd1ddfa644419ccba728132489de : SUCCESS in 4m 29s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/a66d965237a445b398234fc6f118a242 : SUCCESS in 1h 16m 06s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/39449534123b4b609fc2d5fbbac47cd9 : FAILURE in 50m 54s","accounts_in_message":[],"_revision_number":1},{"id":"239db2fc3c29d2dd99c2c3d9409b9a2b65e87341","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-02-09 11:01:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f1bdfc9cafbc4d7153856d4942e7f380422a9de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-09 11:45:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f97e440c014f487db16070fe297ef7ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df7e6c81a1bd4c02ba613562aab3d2fd : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de5a43243b374869855d1c1ff622d94f : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95699b43214a44f0a8b6c7a6a7ad1d63 : SUCCESS in 2m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cfb1bd3a1a1e4a11ac89a3d1481f374d : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1793c5fad40c41549f7a323519f960db : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f70fe892304481b1c5f2682b2ec7b2 : SUCCESS in 3m 03s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/8308e6a1ca5d4092ae99f281fc916654 : SUCCESS in 40m 12s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/0379ffcc136f431080b6a442b8568f37 : SUCCESS in 43m 02s\n- ironic-lib-tox-codespell https://zuul.opendev.org/t/openstack/build/258b365574a049e79105bffeeb126cf1 : SUCCESS in 2m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f02270b5188f9d203d5f57d01eb5e50499d1bdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-09 11:46:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf9e450176835498988a93337e917327dc31e427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-09 13:09:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fafa636cb08d4442822322067a79964e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efacec3ac7d646e489d1c4fa2597666c : SUCCESS in 1m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39a83d1e0bac4c76a2c9f71e0ae2bf61 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98d66ba49b5945999acaed693a1d7760 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0866b0021b44d7a844efd9ebf4de580 : SUCCESS in 4m 10s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/35bb099a60b040a58fce1786f9c8dd9b : SUCCESS in 1h 23m 07s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/aeb36bad9c854bf9b37a347dbdb20ee1 : SUCCESS in 39m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9f70660d14032b7358733936235ec28d13462766","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-09 13:09:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e29f01c16fe9ef7097a8a1d3f9eda61034fbfa35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-09 13:11:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b063c2a413742279fbc9fe1af720429\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2da4eda66e44935a09bfcc00056221d : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d5c9853f1f507cf5ebad665a8c35a945cc52aed","revisions":{"1d5c9853f1f507cf5ebad665a8c35a945cc52aed":{"kind":"REWORK","_number":1,"created":"2024-11-01 00:15:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/933898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/98/933898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/933898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/933898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/933898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/98/933898/1"}}},"commit":{"parents":[{"commit":"b7e12cfc1b89a899c8ca4d10ea47fe8223e98d11","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/b7e12cfc1b89a899c8ca4d10ea47fe8223e98d11"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 09:28:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-01 00:15:04.000000000","tz":0},"subject":"Replace outdated python job template","message":"Replace outdated python job template\n\nWe should use non-versioned template now.\n\nChange-Id: I7385739544ee72de3ffafb4b32611e783f4890b2\n(cherry picked from commit 846baa857cb57850833a6bac52aae2f55e5782fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/1d5c9853f1f507cf5ebad665a8c35a945cc52aed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/1d5c9853f1f507cf5ebad665a8c35a945cc52aed"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
