)]}'
{"id":"openstack%2Fironic-python-agent-builder~1005836","triplet_id":"openstack%2Fironic-python-agent-builder~stable%2F2026.2~Ib5783c66f3d596a97bd68200a5540bd825f7dc7f","project":"openstack/ironic-python-agent-builder","branch":"stable/2026.2","topic":"create-2026.2","hashtags":[],"change_id":"Ib5783c66f3d596a97bd68200a5540bd825f7dc7f","subject":"Update .gitreview for stable/2026.2","status":"MERGED","created":"2026-09-16 08:48:37.000000000","updated":"2026-09-16 09:45:56.000000000","submitted":"2026-09-16 09:44:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005836-create-2026.2","meta_rev_id":"f888e960c76a3a6d884f0d4fd06b918aa641eed4","_number":1005836,"virtual_id_number":1005836,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 09:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2026-09-16 09:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2026-09-16 09:25:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 09:07:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 09:25:49.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"b375467ffdca3794e7cbe52b22f6b2a8c8822d72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-16 08:48:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2a967a6a2f262cb566ae4eedda31b783e4a2943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 09:07:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/926a536a71c841079158cd56a76ad899\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9125aa47a63e453da8a3babc6e47063c : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e0733e0e0f840ee89a9233e818c845c : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f721a38f54cf44e6ac9819ae5a25900b : SUCCESS in 2m 53s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/8352b7ad34c44f56852984cf1b9d4b99 : SUCCESS in 11m 54s\n- ironic-python-agent-check-image-dib-centos9-extra https://zuul.opendev.org/t/openstack/build/5b30a3904ce146cc8e20986a4639e31d : SUCCESS in 13m 16s\n- ironic-python-agent-check-image-dib-centos10 https://zuul.opendev.org/t/openstack/build/f4152eaae0b74142ace87f249b58c307 : SUCCESS in 10m 30s\n- ironic-python-agent-check-image-dib-centos10-extra https://zuul.opendev.org/t/openstack/build/87d616fdddab449bb8576fd44f699f37 : SUCCESS in 10m 10s\n- ironic-python-agent-check-image-dib-debian https://zuul.opendev.org/t/openstack/build/a072fdceb86644768e4697ddb22f4592 : SUCCESS in 8m 20s\n- ironic-python-agent-check-image-dib-debian-arm64 https://zuul.opendev.org/t/openstack/build/1ecaa060f05c4f20b137803a76f70b5f : SUCCESS in 9m 05s\n- ironic-python-agent-check-image-dib-debian-extra https://zuul.opendev.org/t/openstack/build/c86546c789424f11b90409c1c29a399d : SUCCESS in 8m 17s\n- ironic-python-agent-check-image-dib-debian-bookworm https://zuul.opendev.org/t/openstack/build/cb8f4408a82d4a47bff62754743ab34a : SUCCESS in 8m 49s (non-voting)\n- ironic-python-agent-check-image-dib-ubuntu https://zuul.opendev.org/t/openstack/build/27e59e3ef20347a4a6368c7d99badadd : SUCCESS in 8m 12s (non-voting)\n- ironic-python-agent-check-image-dib-ubuntu-arm64 https://zuul.opendev.org/t/openstack/build/c6d2e59bf0264a0ebb83f655416205c7 : SUCCESS in 14m 23s (non-voting)\n- ironic-python-agent-builder-tox-codespell https://zuul.opendev.org/t/openstack/build/9124d38aa2ea4dcbb0d76ebe041dcdc0 : SUCCESS in 2m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8c3d88307526c9a75852d0add96974e95fac44d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-09-16 09:25:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a98d10949938af663bc6410fc8db94c1c83ea94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 09:26:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9377320d7663d01e9eeec6ab4ddbd9371b4e94a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 09:44:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82f4a689771c479b8da79739a1379b54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f56fb459d6d244239c5163bc84695474 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f4dadd154cf4ba097091acd4b16db82 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8809ded193b24a12a92cb4f99dd6a4d3 : SUCCESS in 2m 58s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/58c21a89e14f49d6ad9755096824a8cb : SUCCESS in 12m 26s\n- ironic-python-agent-check-image-dib-centos9-extra https://zuul.opendev.org/t/openstack/build/b26018d27d2f4f80b8ae876821f02100 : SUCCESS in 12m 57s\n- ironic-python-agent-check-image-dib-centos10 https://zuul.opendev.org/t/openstack/build/414c2edaffbf4e1e918eff8a0ccaae1f : SUCCESS in 10m 03s\n- ironic-python-agent-check-image-dib-centos10-extra https://zuul.opendev.org/t/openstack/build/eb95c0bcac1546efaa25ac1172fd009d : SUCCESS in 10m 09s\n- ironic-python-agent-check-image-dib-debian https://zuul.opendev.org/t/openstack/build/48bc2d91f1df4b76867d62a113492d2e : SUCCESS in 4m 03s\n- ironic-python-agent-check-image-dib-debian-arm64 https://zuul.opendev.org/t/openstack/build/1d7e4a367e0e47048e79a635f7ae8308 : SUCCESS in 11m 15s\n- ironic-python-agent-check-image-dib-debian-extra https://zuul.opendev.org/t/openstack/build/9b224ee63d7b4d9a919ce8354e5eeab5 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"fb0cd0fd0ba5304da53b2b083da454cd01efa937","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 09:44:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f888e960c76a3a6d884f0d4fd06b918aa641eed4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 09:45:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4b7ed13a97348ff81393d25b94645de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a44f90297543089f3f4b8a5c971749 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1353723af7d1b0ea0977f8e7d2f68904d4c6878","revisions":{"c1353723af7d1b0ea0977f8e7d2f68904d4c6878":{"kind":"REWORK","_number":1,"created":"2026-09-16 08:48:37.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/36/1005836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent-builder","ref":"refs/changes/36/1005836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent-builder refs/changes/36/1005836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent-builder refs/changes/36/1005836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent-builder refs/changes/36/1005836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent-builder refs/changes/36/1005836/1"}}},"commit":{"parents":[{"commit":"b0ea9c8d71a883b8b96dc4769161ce3a9a1c219a","subject":"Verify registry TLS by default in the podman element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent-builder/commit/b0ea9c8d71a883b8b96dc4769161ce3a9a1c219a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-16 08:48:36.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-16 08:48:36.000000000","tz":0},"subject":"Update .gitreview for stable/2026.2","message":"Update .gitreview for stable/2026.2\n\nChange-Id: Ib5783c66f3d596a97bd68200a5540bd825f7dc7f\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent-builder/commit/c1353723af7d1b0ea0977f8e7d2f68904d4c6878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent-builder/commit/c1353723af7d1b0ea0977f8e7d2f68904d4c6878"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
