)]}'
{"id":"openstack%2Fironic-python-agent~1002971","triplet_id":"openstack%2Fironic-python-agent~master~Ib89e36cdd04efd6fc6f5a06da79db969e1bb422d","project":"openstack/ironic-python-agent","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-16 10:02:49.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-16 10:02:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib89e36cdd04efd6fc6f5a06da79db969e1bb422d","subject":"Replace deprecated pyparsing interface","status":"MERGED","created":"2026-08-30 16:36:40.000000000","updated":"2026-09-16 10:04:05.000000000","submitted":"2026-09-16 10:02:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1002971","meta_rev_id":"f20081ef3fb1f2514d72d17d3a1ae2d9e42d347f","_number":1002971,"virtual_id_number":1002971,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 10:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-09-16 08:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-14 11:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-09-16 08:38:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-30 16:47:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-30 17:35:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-31 09:04:24.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-14 11:26:27.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2026-09-16 08:38:11.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"e10212d3c9af015672871b53e7d5a74a571e29e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-30 16:36:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27154ebb32b29b105b49f859dd23a11e332a0e07","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-30 16:46:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5015a2ce26400887d6f22d5ec7b8c14f0fdf9d92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 16:47:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4304375b08474b2bb31eb54c8db225ee\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a63d95f537046d892a382625feaf7ab : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2819626c19f14f57a28321076fdde20e : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/873b152949fe4cb1a881f732f43daba1 : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e2a485693844455585231927c9fbf2b4ae62023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 17:35:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb1027ab816c4204974f351839b019d5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d88415dfb5764d1b9d20e517f17a89ef : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/332e98db5beb4206949ca5680165c8d7 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea44f40b55b14ae1a1bdbf14b7723038 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/838a7c4660a142959ab16d88fbbfb9d5 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ad4b3ed9c87243dbbdacd44d1ac19535 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b1dc3914514c15b5f88c0f127cad33 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79fce2f52093471b93ca93cccb055950 : SUCCESS in 4m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/acd2c588b26749fbb8c0c25052c097a6 : SUCCESS in 2m 31s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/cdf131e446b9446093698e5c3fc82953 : SUCCESS in 3m 11s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/cea16fb830c44e18a4a3e57f83b5ce44 : SUCCESS in 42m 56s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/cb6460b26efa42148106855849066a42 : SUCCESS in 32m 04s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/b4dd3150c2b44bdfad3c3fd911edfd96 : SUCCESS in 58m 10s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/a7a07a14dacd40769e34fd7222633fcf : SUCCESS in 6m 08s (non-voting)\n- ironic-python-agent-check-image-dib-centos10 https://zuul.opendev.org/t/openstack/build/20ac50d249604eb08498489c2d2d30ab : SUCCESS in 9m 57s (non-voting)\n- ironic-python-agent-check-image-dib-debian https://zuul.opendev.org/t/openstack/build/95ebed975cbc4a84b1151058d2be844b : SUCCESS in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf1bf9f74f2768087a52d5baad56963365f45d6b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-31 09:04:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b59e56dca2fc1b0467e99acb0ec44ced5812b17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 13:23:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8e1b98971ef7af5fa90666e179c679caff613e2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 13:35:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de9277a69eac46a0a3f7f90f10f80095\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dc358119cc7145f0b5d2363293d1f9e3 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/086633306cfa4ad991ce1cb53174f8e0 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b3a78dd7493b4f8d8bdc80ec40184fc6 : SUCCESS in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"775fcebca25f85a98d7c15d603fe53aaa7c27622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:36:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed288c8a53f0451d93dc3707f78fe452\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7efbe6cba8964046851266505a26265e : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e4827e7025f43f8b889e01dc20f4501 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10ef895872d64eb29a8dbc1f4c61f25e : SUCCESS in 8m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/227cbc2b0f844559b3567ad8f45556fb : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4faa2b2d3ae441a8e727407cc382426 : SUCCESS in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c7b8aae81886424a80fc8e36dcc6dd45 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf5b8fb064ac499da7d3ecc042057f36 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/900f950842e141a390d2f47c6c7a5600 : SUCCESS in 6m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/40bbdac315794d92a94c0be14b1952d7 : SUCCESS in 3m 42s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/15fb26d964d94d4ba26294a09521d141 : SUCCESS in 3m 36s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/baf10432c8b5408cbd654d94822f6bd6 : SUCCESS in 51m 04s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/e8339cc2a66d4895a2c70c06da5251a3 : SUCCESS in 1h 05m 33s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/6b038926df4a4a69afdbe77a8b04e149 : SUCCESS in 58m 51s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/deefd83544994da2805878261b613111 : SUCCESS in 11m 53s (non-voting)\n- ironic-python-agent-check-image-dib-centos10 https://zuul.opendev.org/t/openstack/build/94a8a819494f4a5f85411c6234e894ec : SUCCESS in 11m 07s (non-voting)\n- ironic-python-agent-check-image-dib-debian https://zuul.opendev.org/t/openstack/build/7c47addbae8a4541b7c906a04189ab63 : SUCCESS in 7m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"975b076c07cd1f0de6b82db08eac72a2b9b8b5c4","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2026-09-14 11:26:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5e44345e900be1b911ffdca511bc99c8cfa9992f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-09-16 08:38:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"59eeccca6dba4dae21caf3b57a638e074c05a7dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 08:39:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"19dd032c2ef3ce6f9cd39cbaef6db2b96d45d734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 10:02:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef40e3679d7e43a68770587f650e84ed\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/762eb7f4bf23415097a6cba7109d2d76 : SUCCESS in 2m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d67b97e4ed54402a60523885b8cf578 : SUCCESS in 7m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5f0889e7f384ee0982a30cd0577389a : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ffe7fbfbc5cf41a7b5629db3e216d8c3 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea998cbdc064072a68c1d5654b8f759 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc1f6b97245f488b989a457a3a112951 : SUCCESS in 6m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/40fec98db4a4432f95a78aaac0d03128 : SUCCESS in 4m 01s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/41e0379068724453a5ce4f9e66999b2c : SUCCESS in 3m 42s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/e00c53344a5d488fad0877101db2d2dc : SUCCESS in 1h 17m 40s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/4671138346b844ee8b20a0687b32669d : SUCCESS in 1h 14m 13s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/452565e5c9df436dbc98354a6a6ca955 : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":2},{"id":"3731787538843b2c9752d02acf60ae0fdc19c470","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 10:02:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f20081ef3fb1f2514d72d17d3a1ae2d9e42d347f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 10:04:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/735771d225714ef9858a74bae14f9ea5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45fdc6263ace41df87c84b941548c97d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"354fd4abb50a7c9704e8650ccc59d95b409529b3","revisions":{"5571155ff603c0af5decbcb26184b3c922db1a11":{"kind":"REWORK","_number":1,"created":"2026-08-30 16:36:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/1002971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/71/1002971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/71/1002971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/71/1002971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/71/1002971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/71/1002971/1"}}},"commit":{"parents":[{"commit":"3ce553624c20a47a0f1c01cf216db3e1db7e0777","subject":"Merge \"Change md5_enabled default to False\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/3ce553624c20a47a0f1c01cf216db3e1db7e0777"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 16:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 16:36:27.000000000","tz":540},"subject":"Replace deprecated pyparsing interface","message":"Replace deprecated pyparsing interface\n\nparseString was deprecated in favor of parse_string.\n\n[1] https://pyparsing-docs.readthedocs.io/en/latest/whats_new_in_3_0_0.html#api-changes\n* camelCase names have been converted to PEP-8 snake_case names.\n\nChange-Id: Ib89e36cdd04efd6fc6f5a06da79db969e1bb422d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/5571155ff603c0af5decbcb26184b3c922db1a11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/5571155ff603c0af5decbcb26184b3c922db1a11"}]},"branch":"refs/heads/master"},"354fd4abb50a7c9704e8650ccc59d95b409529b3":{"kind":"REWORK","_number":2,"created":"2026-09-07 13:23:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/1002971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/71/1002971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/71/1002971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/71/1002971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/71/1002971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/71/1002971/2"}}},"commit":{"parents":[{"commit":"3ce553624c20a47a0f1c01cf216db3e1db7e0777","subject":"Merge \"Change md5_enabled default to False\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/3ce553624c20a47a0f1c01cf216db3e1db7e0777"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 16:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 13:22:57.000000000","tz":540},"subject":"Replace deprecated pyparsing interface","message":"Replace deprecated pyparsing interface\n\nparseString was deprecated in favor of parse_string.\n\n[1] https://pyparsing-docs.readthedocs.io/en/latest/whats_new_in_3_0_0.html#api-changes\n* camelCase names have been converted to PEP-8 snake_case names.\n\nChange-Id: Ib89e36cdd04efd6fc6f5a06da79db969e1bb422d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/354fd4abb50a7c9704e8650ccc59d95b409529b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/354fd4abb50a7c9704e8650ccc59d95b409529b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
