)]}'
{"id":"openstack%2Fironic-python-agent~718695","triplet_id":"openstack%2Fironic-python-agent~master~I08c73ad14e5b01ebac2490b83997c5452506d4a2","project":"openstack/ironic-python-agent","branch":"master","topic":"more-raid-utils","hashtags":[],"change_id":"I08c73ad14e5b01ebac2490b83997c5452506d4a2","subject":"Move logic to calculate raid sectors to raid_utils","status":"MERGED","created":"2020-04-09 13:04:00.000000000","updated":"2020-04-13 10:52:44.000000000","submitted":"2020-04-13 10:51:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718695-1586775088296-31ef7618","meta_rev_id":"10a833c8ac06c7eee920fddeb55660d5e4c3aa5c","_number":718695,"virtual_id_number":718695,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"date":"2020-04-09 15:27:02.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 10:51:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-09 17:24:11.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2020-04-13 08:51:25.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-12 12:48:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2020-04-13 08:51:25.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 17:24:11.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-12 12:48:01.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-13 08:51:25.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-04-13 10:51:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df87ac4299d2da630e7cd8940f45e3f9a9534346","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-09 13:04:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e124090ef6ab62d75bbe2585789e98d388b08757","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-09 14:42:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3c8dea03234d40f0b52daefb4b26c228 : FAILURE in 1h 36m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e841f377fbb456eb57fca1c3c54ca3c9b23e91bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:59:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaf3f498e50e4b178da53047ef0086a5 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c09bf4e534c44ea9adbd7897334c4362 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d610ef75d13f4bd1aea69e5b80b40a2c : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86f33e98555a41c2b60136c8523272bf : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b07d4e7780b4d76ad00315ca59e0c93 : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/873c36e1030b49fbb294d70d623f6c36 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0660b4d8a5f14cd9b6bfb447c14eca93 : SUCCESS in 8m 23s\n- ipa-tempest-partition-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/a124ae09ff32473da0c526849713997e : SUCCESS in 1h 09m 30s\n- ipa-tempest-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/761b092e387b4a82af0d215fb0da7eea : SUCCESS in 1h 01m 36s\n- ipa-tempest-wholedisk-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/fa339f5bdb45436e8108d552fee1bf0f : SUCCESS in 1h 03m 56s\n- ipa-tempest-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/d56c512db28b42cf8b47077067cf6774 : FAILURE in 42m 39s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/7fa5ae02f0204f16bbd99a1838d7c31a : SUCCESS in 1h 48m 58s\n- metalsmith-integration-ipa-src https://zuul.opendev.org/t/openstack/build/0660486e64b3414985ba61e74fcf2131 : SUCCESS in 1h 38m 22s\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/0e4b099adf024731b63478800605143a : SUCCESS in 18m 13s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/66f5c724035d47ccbd24bdf7f5d2fde2 : SUCCESS in 12m 25s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/500bdb89199f4f0abe50620714f55ab7 : SUCCESS in 1h 51m 25s (non-voting)\n- ipa-tempest-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/ddb92c27b6284bf19312f01d275369a4 : SUCCESS in 1h 07m 48s (non-voting)\n- ipa-tempest-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/7f74188a8f7e40d09c9f26784cf2207e : SUCCESS in 1h 06m 00s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6e9b9445c460489ea7d875c65dae89cc : SUCCESS in 6m 12s (non-voting)\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/c91cf5036b754188bd9b08c6c34eb3b1 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b23b893174ef153af6e33a3dc3afd7c8cf7b592","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-09 15:27:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"043c01ea3bbda6f9f69cebba2dc33e2f0e2b9ab9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-09 17:24:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9e4f09844253455d979855b3233e01bf : FAILURE in 1h 31m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e0a2e58d491c14f6f7b09fb339eb8245587935d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 17:26:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0462fc5c7a4e48fbbffd0dbca157706c : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e154638a18254effb53cbad9c63b8998 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9988e22ed7443f2991d84cf490764e7 : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb66130cbc99438b9087ca7e66699920 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/709049d82c35430c920ba7139819c280 : SUCCESS in 8m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60d909b794d9436181c787eed01cf7fe : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5287e81ec2684384910982d8b3926f9e : SUCCESS in 9m 35s\n- ipa-tempest-partition-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/9e68817d747746758d8251028a686754 : SUCCESS in 1h 05m 27s\n- ipa-tempest-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/866f9647b5b14619932bc0738f2d7119 : SUCCESS in 1h 02m 00s\n- ipa-tempest-wholedisk-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/4237425f38554f19b4e660e23a95a378 : SUCCESS in 1h 25m 35s\n- ipa-tempest-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/7bac9387f09e417cb12c4f1ea00c9779 : SUCCESS in 1h 20m 48s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/7dff48985a3d43cfb7c9db9e54c43c41 : SUCCESS in 1h 50m 04s\n- metalsmith-integration-ipa-src https://zuul.opendev.org/t/openstack/build/4a5fd545735544d688943acd7a000df6 : SUCCESS in 1h 39m 38s\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/1ca9e10b94554fca90a90cf2a81cb7c0 : SUCCESS in 20m 16s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/572bce69c768486c90648f583b5dcb34 : SUCCESS in 11m 19s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/f9d1a55e093345e895268ed96ee4a13a : SUCCESS in 1h 34m 04s (non-voting)\n- ipa-tempest-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/37dc65a2aa3b4da7b426431f6dca3cd6 : SUCCESS in 1h 04m 22s (non-voting)\n- ipa-tempest-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/b40d02a76bef44c284523bc64fe13472 : SUCCESS in 1h 19m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/406dbf030671438a9fa2afde797de941 : SUCCESS in 7m 06s (non-voting)\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/f512acd6e08b4cb683e85f406674d5f0 : SUCCESS in 7m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d691f2e7685f1af6fd2d48bce8591ba99f7d28dd","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-12 12:48:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1a258759e44252cde8d472455f175a4bc2ce5fea","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-04-13 08:51:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you Riccardo.","accounts_in_message":[],"_revision_number":1},{"id":"9473a5f6b6804da1de4b1b709e2369cca5261b57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 08:51:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7f4c96e18bb39d013b7a605aae74fb6fd584efc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 10:51:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9c2a70ffb317964b254b6448a1267149c724d22b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 10:51:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce211c1170974fc8b1b54ae23d0092f9 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/250477d465e140b489d2c3ec4e09e16e : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8617a9750f344dd68a2a5ced0833355b : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48cefbf9374b45d4b68c4f0bc202f7ec : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8418e6c9ba2d4cb19525aded5384fe8c : SUCCESS in 5m 41s\n- ipa-tempest-partition-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/d5e3d14cd0dc43b09a06607a0288851d : SUCCESS in 1h 08m 08s\n- ipa-tempest-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/f2a9daad321d46a5b15b2c933936cc92 : SUCCESS in 1h 15m 48s\n- ipa-tempest-wholedisk-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/2221fbb2abc64b978733ee0625facae0 : SUCCESS in 51m 02s\n- ipa-tempest-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/749215ee51be410297b933b8f4d00f7a : SUCCESS in 1h 16m 43s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/625fdfcef3a8469cbac86a8f8a352ad7 : SUCCESS in 1h 20m 56s\n- metalsmith-integration-ipa-src https://zuul.opendev.org/t/openstack/build/0652edc9df3244b2a8ea197dc819f981 : SUCCESS in 1h 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"3dfbeba820fd99935264c987e296b38bc4f39743","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 10:52:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3369cfc4dc47e7a1c8e5aa45bf7fe2 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3966871f47f9d2acb209647c618dd824abc27807","revisions":{"3966871f47f9d2acb209647c618dd824abc27807":{"kind":"REWORK","_number":1,"created":"2020-04-09 13:04:00.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/95/718695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/95/718695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/95/718695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/95/718695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/95/718695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/95/718695/1"}}},"commit":{"parents":[{"commit":"83b5a8b20269645dee925cb87c91387396f35274","subject":"Merge \"Move logic for raid start sector to raid_utils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/83b5a8b20269645dee925cb87c91387396f35274"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 12:58:42.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 13:03:37.000000000","tz":120},"subject":"Move logic to calculate raid sectors to raid_utils","message":"Move logic to calculate raid sectors to raid_utils\n\nSome more raid related logic moved to raid_utils.\n\nChange-Id: I08c73ad14e5b01ebac2490b83997c5452506d4a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/3966871f47f9d2acb209647c618dd824abc27807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/3966871f47f9d2acb209647c618dd824abc27807"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
