)]}'
{"id":"openstack%2Fironic-python-agent~778201","triplet_id":"openstack%2Fironic-python-agent~master~Iedff5325e0737556d5eb3da601c984ddfc633873","project":"openstack/ironic-python-agent","branch":"master","topic":"default-execute","hashtags":[],"change_id":"Iedff5325e0737556d5eb3da601c984ddfc633873","subject":"Remove default parameter from execute","status":"MERGED","created":"2021-03-02 15:19:57.000000000","updated":"2021-03-03 07:15:59.000000000","submitted":"2021-03-03 07:14:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778201-default-execute","meta_rev_id":"3d41b3e5a6400c2093e231b241bd276efdd269e6","_number":778201,"virtual_id_number":778201,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-03 07:14:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2021-03-03 05:51:05.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2021-03-02 16:12:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2021-03-03 05:51:05.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 16:12:10.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2021-03-02 17:24:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 05:51:05.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"}],"messages":[{"id":"2be01ea014d4cfc977e6a4a2dd81645bfd2dc364","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-03-02 15:19:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ee0bdb2b09d9fccb269ede9462aec3c6ac7d168","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2021-03-02 16:12:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"66903483f661e7543525ad844782c906c0450542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 17:24:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d7366d0e1b54964bff72d81e3ca7996 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71766e2a82f74917890c213840187356 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41a25a2eb15a439696080acf49144878 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9179ef48aea4ac0b781e520f57ac3e7 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adfa29f337fd4ed287f384387459c60b : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36be2fe9244546719becc2071cc22c36 : SUCCESS in 7m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/08a3d8f64fa844a292dd4f75fc4b3fe3 : SUCCESS in 5m 29s\n- ipa-tox-examples https://zuul.opendev.org/t/openstack/build/38cd67de0e75486c844ecdf3bf5d9083 : SUCCESS in 5m 52s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/94766a50a09a4594bee4e86e50a10f24 : SUCCESS in 1h 23m 54s\n- ipa-tempest-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/9588158d3b474d95924f32fa58e6822f : SUCCESS in 1h 13m 12s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/b0680eb748b6427390064c395fb18f2e : SUCCESS in 1h 34m 17s (non-voting)\n- metalsmith-integration-ipa-src https://zuul.opendev.org/t/openstack/build/db5e1f501e6b46c28657f525cf42a05b : SUCCESS in 1h 13m 09s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/ba5aa811f0b947b7965b1d98df0f0957 : SUCCESS in 1h 32m 55s (non-voting)\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/8d71c61dc9724c2c910f27ee704c450e : SUCCESS in 15m 27s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/0cf724d23ae549fbbddeb220ead85579 : SUCCESS in 11m 00s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/267d57ed88eb4c7180fbc411d1de81d9 : SUCCESS in 2h 01m 11s (non-voting)\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/9865de50eefe473299ccc7ec96b435a1 : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5756441d2d2bb57a6a5769a183cbdce94c918cea","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2021-03-03 05:51:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you Riccardo.","accounts_in_message":[],"_revision_number":1},{"id":"2800a2d19247357b0cb28199cd7dd8b4b451a581","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 05:51:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a59824a4ce83f87bbc138bcb726cbed7a8202c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 07:14:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3f296ab0ae047b583eb067eb39345b1 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cd55cabfb1942c2a76434c48a53b820 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d74cb41d4914ffcb24f2288a5a5fb15 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a49bbe5c0346bdb2cd899e2f4ac350 : SUCCESS in 8m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/44e41c978fc8418691c3eb7fdf4b4f2f : SUCCESS in 5m 09s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/982123dde722409cb0bde3f8dbaf9cbb : SUCCESS in 1h 08m 13s\n- ipa-tempest-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/4b19e9b7b7b14e02ace4e9fb731e6033 : SUCCESS in 1h 21m 23s\n- metalsmith-integration-ipa-src https://zuul.opendev.org/t/openstack/build/973f60e504e4466d8b63fe30a0a61d18 : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e22d58a3b8ff88fd98200bbc31800e1ea26e6c60","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 07:14:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d41b3e5a6400c2093e231b241bd276efdd269e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 07:15:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab09f82f32574f22b2a64f0d0aa48a6e : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bff252c7263c70edb52644f7ac64d39b9f6b0bd0","revisions":{"bff252c7263c70edb52644f7ac64d39b9f6b0bd0":{"kind":"REWORK","_number":1,"created":"2021-03-02 15:19:57.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/01/778201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/01/778201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/01/778201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/01/778201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/01/778201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/01/778201/1"}}},"commit":{"parents":[{"commit":"4354d4f792f188e0726dea851ff59fd02047de37","subject":"Merge \"Remove nvme-cli warning and delay on nvme-format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/4354d4f792f188e0726dea851ff59fd02047de37"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-03-02 15:19:32.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-03-02 15:19:32.000000000","tz":60},"subject":"Remove default parameter from execute","message":"Remove default parameter from execute\n\nThe param check_exit_code from the processutils extension execute has\ndefault already at [0]\nSee:\nhttps://opendev.org/openstack/oslo.concurrency/src/branch/master/oslo_concurrency/processutils.py#L214\n\nChange-Id: Iedff5325e0737556d5eb3da601c984ddfc633873\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/bff252c7263c70edb52644f7ac64d39b9f6b0bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/bff252c7263c70edb52644f7ac64d39b9f6b0bd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
