)]}'
{"id":"openstack%2Fironic-python-agent~793082","triplet_id":"openstack%2Fironic-python-agent~master~Ibe02786609aa0de65115897d8f4a9b4f36c8aed2","project":"openstack/ironic-python-agent","branch":"master","topic":"story/2008923","hashtags":[],"change_id":"Ibe02786609aa0de65115897d8f4a9b4f36c8aed2","subject":"Make _get_efi_bootloaders return relative paths","status":"MERGED","created":"2021-05-26 01:14:37.000000000","updated":"2021-05-27 13:11:24.000000000","submitted":"2021-05-27 13:09:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793082-story/2008923","meta_rev_id":"4062cede98ef614cc98ac312bb7f5b9903dd3629","_number":793082,"virtual_id_number":793082,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 13:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-05-27 08:11:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-27 11:39:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-27 11:39:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-26 02:44:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-26 08:06:27.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2021-05-26 20:00:54.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"},{"updated":"2021-05-27 08:11:13.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-05-27 11:39:12.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"a62f2bee1c9048edca3d07f5f9124d6fe30f4ff6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-05-26 01:14:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26f38db6745453257796b6bde20ff13e6145b162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 02:44:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9a2ef46139b46ef88bd816c974e9abc : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3a599b1744746b2a2db59713d8766f9 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da70dd0661764914beb90a709d5f9972 : SUCCESS in 7m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/383e7f388fa540de9b354c37505fe1f5 : FAILURE in 5m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e94956f8b85a4f81ba0547aa6e79ebb7 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfcc130cdbd9487885ed689131fd8c10 : SUCCESS in 22m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bb0afc213f124ca19953aa74d74e0384 : SUCCESS in 8m 14s\n- ipa-tox-examples https://zuul.opendev.org/t/openstack/build/e739670caf6c4a449c69f0ab2654bd96 : SUCCESS in 8m 08s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/297dfec9a1334592add2aab1cd3d0b4f : SUCCESS in 1h 24m 10s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/e2afc8308d0d4a078db5148f8dff7b3b : SUCCESS in 1h 20m 52s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/bdca2155680a4a43850117f2dc463ea2 : SUCCESS in 1h 03m 44s (non-voting)\n- metalsmith-integration-ipa-src https://zuul.opendev.org/t/openstack/build/0e8ca3a2b84e444aaa9cee456b1ee478 : FAILURE in 18m 48s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/d26f08feaf4f4d74873f0dcb1b84e874 : SUCCESS in 1h 26m 14s (non-voting)\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/e103bd5ce5ab4551a368b84a66ba9622 : SUCCESS in 19m 00s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/483e4be5ca3a48609bce2e8541e69997 : SUCCESS in 12m 52s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/0a21c50da3604824b997215a29fe6659 : POST_FAILURE in 16m 04s (non-voting)\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/05c4bc667c8d45acb4f509262fb08e0f : SUCCESS in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc1258e60b9468693ccf51ff80422867bbddb9da","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-05-26 08:06:27.000000000","message":"Patch Set 1:\n\nrecheck missing python lib protobuf\u003d\u003d\u003d3.17.0 but it should be there","accounts_in_message":[],"_revision_number":1},{"id":"f45447a88bc23dea2337fe9255ccb395c5ce02e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 10:26:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c83ba6bba3ce4901a71d5e8d94b0c5bb : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea3f2054b11e48e38430dde2368520ea : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dc7277e4cff447d97a88ed268021eec : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbba077ad62445c4a94e4a2fef320070 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a3af0ce0dc54b34a97cc523c865e1b2 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a35c83647ae4eb2ac38733ae867ac62 : SUCCESS in 6m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e69e9ec7943547fbbb9e574020523e1a : SUCCESS in 6m 32s\n- ipa-tox-examples https://zuul.opendev.org/t/openstack/build/a387ff91ca664fc594102a76af669cb4 : SUCCESS in 5m 15s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/bb86253ef6134122afba7b38a0d4d945 : SUCCESS in 1h 16m 48s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/9b8761942f084172b4cd7643d1da6bb2 : SUCCESS in 1h 25m 20s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/80b43b97e9a7421b8c241246bbb083f3 : SUCCESS in 1h 02m 01s (non-voting)\n- metalsmith-integration-ipa-src https://zuul.opendev.org/t/openstack/build/d5843123ff3f405bae85d3f436656f9b : SUCCESS in 1h 25m 04s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/d47008ae71454df6840635f4679d31c8 : SUCCESS in 1h 21m 52s (non-voting)\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/909479e35ad443e29558adac18a10631 : SUCCESS in 16m 17s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/f6c9a82c880f4d6e8361b5b557412b77 : SUCCESS in 12m 26s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/7c8183fd5bb94dd7a630006405aaaed2 : SUCCESS in 2h 16m 06s (non-voting)\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/dfc720e32b1a497ba578dae989f9853c : SUCCESS in 3m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43a11255722b39ab01b5a1d1e7851817bd927d47","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-05-27 08:11:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04627bfcf535e74cec01d3c36c7bd352d1d09c0c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-05-27 11:39:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bcdede8cae2b40d804c8c60c0961f7b8e1af010d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:39:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6bb26d51ceacd3f533853b2c717d7960bd378d3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 13:09:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708e474018414d08aa234e4947f22938 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8b58432788247e4a474b34b6ec68ae3 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/568ce18a35774396b76ba978b5baecf3 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365dc38d9e9f43778f1e583fae050cf7 : SUCCESS in 7m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/41920b1200bc46338132a7002b3f730c : SUCCESS in 5m 12s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/2bb1a624c92d40f5bd7af9ecc2f73936 : SUCCESS in 1h 14m 55s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/f0eb74ca09964a30b6e521e2a997a9b1 : SUCCESS in 1h 26m 28s\n- metalsmith-integration-ipa-src https://zuul.opendev.org/t/openstack/build/3e27bca66cb946c1b53a3547a2a07169 : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ff6a210cf16a7846b2c85ef418352bfab6eaf842","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 13:09:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4062cede98ef614cc98ac312bb7f5b9903dd3629","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 13:11:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d10ed4a8247496bba8d641b61224b69 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10d18c41136cc645ee99d41acfb6031b9158e1fb","revisions":{"10d18c41136cc645ee99d41acfb6031b9158e1fb":{"kind":"REWORK","_number":1,"created":"2021-05-26 01:14:37.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/82/793082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/82/793082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/82/793082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/82/793082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/82/793082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/82/793082/1"}}},"commit":{"parents":[{"commit":"b4dd03168e631baeb0edd2c4269e1258a8fed839","subject":"Merge \"Enable out-of-order writes when writing whole disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/b4dd03168e631baeb0edd2c4269e1258a8fed839"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-05-25 22:41:40.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-05-25 23:21:15.000000000","tz":720},"subject":"Make _get_efi_bootloaders return relative paths","message":"Make _get_efi_bootloaders return relative paths\n\nTo make this function useful for purposes other than efibootmgr\nentries, this change moves the path manipulation to _run_efibootmgr.\n\nThis change also adds boot*.efi entries to BOOTLOADERS_EFI so that it\nincludes every entry in the UEFI Spec 2.9[1] Table 3-2 UEFI Image\nTypes.\n\n[1] https://uefi.org/sites/default/files/resources/UEFI_Spec_2_9_2021_03_18.pdf\nStory: 2008923\nTask: 42521\n\nChange-Id: Ibe02786609aa0de65115897d8f4a9b4f36c8aed2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/10d18c41136cc645ee99d41acfb6031b9158e1fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/10d18c41136cc645ee99d41acfb6031b9158e1fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
