)]}'
{"id":"openstack%2Fironic-python-agent~816941","triplet_id":"openstack%2Fironic-python-agent~master~If5da3075e813ae01df3decf8f0647aba111b0515","project":"openstack/ironic-python-agent","branch":"master","topic":"partx_a","attention_set":{},"removed_from_attention_set":{"6618":{"account":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"last_update":"2021-11-08 20:58:57.000000000","reason":"Change was submitted"},"11292":{"account":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"last_update":"2021-11-08 16:42:50.000000000","reason":"Arne Wiebalck replied on the change"}},"hashtags":["ironic-week-prio"],"change_id":"If5da3075e813ae01df3decf8f0647aba111b0515","subject":"Re-read the partition table with partx -a","status":"MERGED","created":"2021-11-08 07:22:52.000000000","updated":"2021-11-08 21:01:41.000000000","submitted":"2021-11-08 20:58:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"816941-partx_a","meta_rev_id":"775c4fcbda3f93a5d1d71311380a3e496164c57f","_number":816941,"virtual_id_number":816941,"owner":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-08 20:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"all":[{"value":2,"date":"2021-11-08 16:09:47.000000000","_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":2,"date":"2021-11-08 16:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":1,"date":"2021-11-08 16:40:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"all":[{"value":1,"date":"2021-11-08 16:09:47.000000000","_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-08 14:36:04.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2021-11-08 16:09:47.000000000","updated_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"reviewer":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"state":"REVIEWER"},{"updated":"2021-11-08 16:35:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-08 16:40:12.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"6d9e2a3cc695905f8462bf7a2c86345a4807d4ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-11-08 07:22:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b71758337b5d1e87fb13cb1b4d62b96a5433186a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-11-08 14:36:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41b41d87b07dc0fce7727fe810bd5ca1a2375ee7","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-11-08 15:11:58.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"e9d600420687796c05a5e74915966d5a22f6d5d1","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2021-11-08 16:09:47.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8c971d8206444897b6aae29a94391e2e3c3c4f69","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2021-11-08 16:10:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18f64775a4378eeae99a5f88694f48e95a64a89c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 16:35:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02d9dafd6ade4f1f823267cab4d63533 : SUCCESS in 7m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83b4ed3721be4b26b06e11c3ae7fdd0a : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e5470e7ef7a494283d323cc54574217 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c501f87c8c80404c890bc37c21e3a378 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4074f31b21240af8ce559a4d05dc70a : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fdbb73d4ad0441fbd891ae18ddf2f88 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce2658e0b1a47f8ae99fd266cd84648 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/782a91a2162a4dc09aa4d437896916d2 : SUCCESS in 3m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6eff971499d94927bd096f13cfcf5586 : SUCCESS in 4m 26s\n- ipa-tox-examples https://zuul.opendev.org/t/openstack/build/2d278137915e43e28989be79a3ce4083 : SUCCESS in 6m 51s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/e1755971416746afa2abc186ada5a20b : SUCCESS in 1h 03m 06s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/fa2d224d6c61429395b4cff715e6eecd : SUCCESS in 1h 00m 19s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/235a01f4b8a541d89f110f2dcbf6ffb0 : FAILURE in 1h 55m 35s (non-voting)\n- metalsmith-integration-ipa-src https://zuul.opendev.org/t/openstack/build/7e198b3cba574924abf1c458bf5c63aa : SUCCESS in 58m 58s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/0a613d9f9e2641a4b120317b05cb6eae : SUCCESS in 1h 09m 51s (non-voting)\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/b69af1fbe78a47e89cefda098ee38a2f : SUCCESS in 16m 06s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/151c5ad491b143b198ead229a8b662cb : SUCCESS in 18m 29s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/7294409df9bf4b2c86e5c9ecf8dbcda7 : FAILURE in 1h 32m 13s (non-voting)\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/141d3cc34fd347428f259396873869db : SUCCESS in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08b9b4c6acb04e8e4085334f1ccf708c2bc3fb45","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-11-08 16:40:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f477e3ddd762f8a2dde1d39d8f11dd41071c671d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 16:41:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61fcabe96f9406a0909147ffa0f0ed6378efaf34","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-11-08 16:42:50.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0c0882b20bc26782c30db922f6d7545336343883","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-11-08 16:43:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90e2d9f67ed9662ee1d211592089689a81dcab88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 18:36:46.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5eabd468dd29799c1af7814b1056ad79b374eaff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 20:58:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/833d87a77d6c4e02a54e87ca6eb64450 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cb71fa1386a4b299365352135fed541 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0958df900b0141f7adabec6c7a79ba39 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6148b9118ec47cfba6278fa3f60b920 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e619118caca41ea8bd5d54a6fd79980 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c332817fadc43989cfbfcd0db100f6f : SUCCESS in 5m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/893461f9f44a469786aa41e9c2abaada : SUCCESS in 7m 38s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/5f7b161cfe6846fa9d8d1fbc7c18576f : SUCCESS in 1h 29m 14s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/35d7b8e007a248718dba2336b644d0bc : SUCCESS in 1h 24m 10s\n- metalsmith-integration-ipa-src https://zuul.opendev.org/t/openstack/build/40d35f66649d437aa507eeace9993b61 : SUCCESS in 1h 24m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8976a3119a8bb96ee9b97d3f090a9a3713f0f266","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 20:58:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"775c4fcbda3f93a5d1d71311380a3e496164c57f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:01:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fd489259ec24b279e594b8cabef9673 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49b7d6e7758f4f6fb19c624cebdb5c58 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc8c1f16f9a00e2bff21612d1a9cf0ea0f3addf0","revisions":{"dc8c1f16f9a00e2bff21612d1a9cf0ea0f3addf0":{"kind":"REWORK","_number":1,"created":"2021-11-08 07:22:52.000000000","uploader":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"ref":"refs/changes/41/816941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/41/816941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/41/816941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/41/816941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/41/816941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/41/816941/1"}}},"commit":{"parents":[{"commit":"0b56cca7f0116a99ac1dd3e5d5aed5647aad662d","subject":"Merge \"Fix UEFI record regex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/0b56cca7f0116a99ac1dd3e5d5aed5647aad662d"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2021-11-06 12:28:06.000000000","tz":60},"committer":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2021-11-06 12:43:48.000000000","tz":60},"subject":"Re-read the partition table with partx -a","message":"Re-read the partition table with partx -a\n\nRe-read the partition table with \u0027partx -a\u0027, rather than \u0027partx -u\u0027.\n\nThis should fix an timing issue where the bootloader installation\nfails to mount the EFI partition from a whole disk image since it\nis not yet aware of the new partitions (observed with both, the\niscsi and the direct deploy interface).\n\nChange-Id: If5da3075e813ae01df3decf8f0647aba111b0515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/dc8c1f16f9a00e2bff21612d1a9cf0ea0f3addf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/dc8c1f16f9a00e2bff21612d1a9cf0ea0f3addf0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
