)]}'
{"id":"openstack%2Fironic-python-agent~913202","triplet_id":"openstack%2Fironic-python-agent~master~I40ab2b3a17b255cc05b1054d4ec7ac788e2debb9","project":"openstack/ironic-python-agent","branch":"master","topic":"bug/2057668_wip","hashtags":[],"change_id":"I40ab2b3a17b255cc05b1054d4ec7ac788e2debb9","subject":"Add get_additional_skip_list and get_skip_list","status":"ABANDONED","created":"2024-03-14 17:18:51.000000000","updated":"2024-03-27 16:56:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac6f570f6a54ccac3e433b584151aa9a6986caec","_number":913202,"virtual_id_number":913202,"owner":{"_account_id":31031,"name":"Daniel King","email":"openinfra@anewmind.me","username":"DanKingTech"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-14 20:27:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 17:24:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 20:27:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"938b6f730061e3168c629af4f18d9b8718c7b472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31031,"name":"Daniel King","email":"openinfra@anewmind.me","username":"DanKingTech"},"date":"2024-03-14 17:18:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"686fe787322f796edeeee87316f2943555491f95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 17:24:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b5e87e5f9a84bfba71beb58f0b2df26\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b385798d92a4943976056154afe917b : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8dcc43e2ae0b4aab87463894f4d9b873 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/369fb6928ba543e3b80726a4c4f15b0c : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9728fc3e874ff76ba160e3a338c18faa84518e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 20:27:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73b92ed592984016896a71ed7a43928d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75578bfc7ef149d3ae621ee78ab35594 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ffb7222a2c94de79a1f07b072b20e65 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c008be9957ff4591ae03a351e3b7dc66 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79f91fa07f354e70ad628b674eabeeef : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6519e246b11a4002823b693859e6a8fb : SUCCESS in 6m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9bf769629f854cbb90a407be970687f1 : SUCCESS in 3m 31s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/e9f5034e03a5426c97ff6a06a9059d67 : SUCCESS in 3m 30s\n- ipa-tox-examples https://zuul.opendev.org/t/openstack/build/4a61b942e0d24e7d8e3ff0a53d893796 : SUCCESS in 3m 19s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/c8995cbc17214e3bb119af3c3c576731 : SUCCESS in 1h 41m 37s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/f658ca1d3ece49f7964d24002ad6f84d : SUCCESS in 1h 33m 06s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/43ec930c18704f57bf40a3aef93a7a26 : SUCCESS in 1h 14m 53s\n- metalsmith-integration-ipa-src-legacy https://zuul.opendev.org/t/openstack/build/36b7a6ac13fc4928be97680d5850a188 : FAILURE in 1h 06m 28s (non-voting)\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/dff849163c594822aeb0b1314ee0f70b : SUCCESS in 1h 21m 26s\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/bd539dbeffb4400787d99732dfae5d6e : SUCCESS in 16m 18s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/690548495b6144e3aca833e98d792d6c : SUCCESS in 11m 58s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/1d694e18b75240cca47a034a4657a91d : TIMED_OUT in 3h 02m 51s (non-voting)\n- ipa-tox-codespell https://zuul.opendev.org/t/openstack/build/06fa4eaf11534b98958517f66ab83cf4 : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac6f570f6a54ccac3e433b584151aa9a6986caec","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31031,"name":"Daniel King","email":"openinfra@anewmind.me","username":"DanKingTech"},"date":"2024-03-27 16:56:33.000000000","message":"Abandoned\n\nThis is being handled instead by 913209","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e493492ca35386636cd63b3bcd836bd0f2049007","revisions":{"e493492ca35386636cd63b3bcd836bd0f2049007":{"kind":"REWORK","_number":1,"created":"2024-03-14 17:18:51.000000000","uploader":{"_account_id":31031,"name":"Daniel King","email":"openinfra@anewmind.me","username":"DanKingTech"},"ref":"refs/changes/02/913202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/02/913202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/02/913202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/02/913202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/02/913202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/02/913202/1"}}},"commit":{"parents":[{"commit":"815e1f462f9cac0473699f06753e0850789061af","subject":"Merge \"reno: Update master for unmaintained/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/815e1f462f9cac0473699f06753e0850789061af"}]}],"author":{"name":"Daniel King","email":"openinfra@anewmind.me","date":"2024-03-13 20:28:34.000000000","tz":-240},"committer":{"name":"Daniel King","email":"openinfra@anewmind.me","date":"2024-03-14 17:15:44.000000000","tz":-240},"subject":"Add get_additional_skip_list and get_skip_list","message":"Add get_additional_skip_list and get_skip_list\n\nAdd the abstract method get_additional_skip_list to HardwareManager as a\nnew hood and add the get_skip_list method to GenericHardwareManager\nwhich combines get_skip_list_from_node and the new method, the later of\nwhich is called with dispatch_to_manager to allow it to be implemented\nby custom hardware managers.\n\nRelated-Bug: 2057668\nChange-Id: I40ab2b3a17b255cc05b1054d4ec7ac788e2debb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/e493492ca35386636cd63b3bcd836bd0f2049007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/e493492ca35386636cd63b3bcd836bd0f2049007"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
