)]}'
{"id":"openstack%2Fironic-python-agent~936679","triplet_id":"openstack%2Fironic-python-agent~unmaintained%2F2023.1~I1803833ba1282358f463b738fe9057000622b279","project":"openstack/ironic-python-agent","branch":"unmaintained/2023.1","topic":"create-2023.1","hashtags":[],"change_id":"I1803833ba1282358f463b738fe9057000622b279","subject":"Update .gitreview for unmaintained/2023.1","status":"MERGED","created":"2024-11-29 07:54:13.000000000","updated":"2025-03-19 20:40:35.000000000","submitted":"2025-03-19 20:39:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936679-create-2023.1","meta_rev_id":"209ae9bcb3999b63679377e9c9d0a96a28f28f6e","_number":936679,"virtual_id_number":936679,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-19 20:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2024-11-29 08:17:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2024-11-29 08:17:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-29 08:17:27.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-11-29 08:25:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-19 17:21:07.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"}],"messages":[{"id":"6417bf6f6b60c3cde681749d25d7cb331a08ea3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-11-29 07:54:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da3ef8447d5a7222a583557a14ae4aa904cfbd5e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-11-29 08:17:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa8ac6a903ae34e414ddc740009b220c84427f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:25:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/545b2123fa51414696ac8dfd22aa975d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1c5e13c88bd42a283e36db096baab4f : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a970ecc1ff7a4c75bf73764bd46a7917 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a028dd075f401a971050c8757417e2 : SUCCESS in 6m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/afe2009aabfe486496311b8f210438eb : SUCCESS in 3m 21s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/649fbf7c003b44829bf0ca7da1ed263d : FAILURE in 20m 05s\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/5305dcc5c23a49fb8b6e18b88ec320e3 : SUCCESS in 14m 44s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/d04dcdf948594bf4a03e8b3f565704d6 : SUCCESS in 10m 32s (non-voting)\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/6953488b44224235bf7b4eecaee5bd5b : FAILURE in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7f5ca48c8e751c39363ad57ae1e4a3c2e1118ac","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-03-19 17:21:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a94399b5389226b112f27f6719a993d1e2c6142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 19:02:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f2e8edb741d4c9bb1d84ad4521f89c8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c388c12374fd4d968d5b0b958875ae3c : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/affe1dd3d85b4d9881a4b67d743dc550 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35b62823b4846d290b53d69b94abcd8 : SUCCESS in 13m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9214733e7eee43b39abedb183062e8c9 : SUCCESS in 3m 46s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/ca9666135a924f4eb6b4766c9a5b9feb : SUCCESS in 1h 02m 28s\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/09f877f4b8984f3ead0858afd520ca40 : SUCCESS in 21m 32s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/74878eff924047d49c7e0e32d93afee9 : SUCCESS in 9m 36s (non-voting)\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/d8fbea01bc814c77bfa15689c0659823 : FAILURE in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f95c7cbdf8ead0cca9e424f7a036239290e3cf80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 19:02:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"471e5e11a1de7f92fc3ccf768adacea98692fcd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 20:39:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e42ea37f8d24b08ab19c6ceafa7b8d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8039dfb6783a4935a2540acb68abad36 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f98b587b27347a8b9ba74ab32587943 : SUCCESS in 5m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6b53509622a04e8e82234c569a0a023f : SUCCESS in 3m 36s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/c51b32de475045df9ca7ed0ca9776be3 : SUCCESS in 1h 31m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3569f913d60bc5256fe7e8bd1f73937f360076cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 20:39:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"209ae9bcb3999b63679377e9c9d0a96a28f28f6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 20:40:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54722f23352841018975dc82848ff2a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e52f6ea0dfa4421a24b4fd9abf77539 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4cf1abcccbd3991d00a8ff1a51c3cc103e9f9ad5","revisions":{"4cf1abcccbd3991d00a8ff1a51c3cc103e9f9ad5":{"kind":"REWORK","_number":1,"created":"2024-11-29 07:54:13.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/79/936679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/79/936679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/79/936679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/79/936679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/79/936679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/79/936679/1"}}},"commit":{"parents":[{"commit":"a488be6e21b036e13ea567ed334707ad09901d88","subject":"Remove non RE2 job config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/a488be6e21b036e13ea567ed334707ad09901d88"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-11-29 07:54:06.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-11-29 07:54:06.000000000","tz":0},"subject":"Update .gitreview for unmaintained/2023.1","message":"Update .gitreview for unmaintained/2023.1\n\nChange-Id: I1803833ba1282358f463b738fe9057000622b279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/4cf1abcccbd3991d00a8ff1a51c3cc103e9f9ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/4cf1abcccbd3991d00a8ff1a51c3cc103e9f9ad5"}]},"branch":"refs/heads/unmaintained/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
