)]}'
{"id":"openstack%2Fironic-python-agent~946580","triplet_id":"openstack%2Fironic-python-agent~master~I2b6d542ce7af3eeeff23f06e43e82de5d7b09701","project":"openstack/ironic-python-agent","branch":"master","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-04-18 00:53:34.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I2b6d542ce7af3eeeff23f06e43e82de5d7b09701","subject":"Base standalone job on redfish standalone","status":"MERGED","created":"2025-04-07 18:12:40.000000000","updated":"2025-04-18 00:54:53.000000000","submitted":"2025-04-18 00:53:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946580","meta_rev_id":"857aa77bf6fb8ca22695c04a1b824cb7650e4fbf","_number":946580,"virtual_id_number":946580,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-18 00:53:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"all":[{"value":2,"date":"2025-04-17 20:02:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":2,"date":"2025-04-16 08:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-11 15:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":1,"date":"2025-04-16 08:50:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-07 18:22:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-07 20:43:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-11 15:18:47.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2025-04-16 08:50:48.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-04-17 20:02:35.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"}],"messages":[{"id":"aab0b2056d25cf20319459e56d96c240141c556b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-04-07 18:12:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16b2401d06cce3a3ffa44ebd88ec9c23527fdd9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 18:22:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c7b734d46e246fdbe965258f1ee1a2e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/407b763f639d41da95cdcaee5b99ede7 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a8ffda1029834bd59cf038a632fd715f : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9013d962dc2d4618a169b6e30116079f : SUCCESS in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0c3568a4d613e1ac1d72026320f996f130dedeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 20:43:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd536883828b4871b7245a0f989eebf0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83a7bf037d364804845fa4a70a82b1ab : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6df353d61b041bd9ecb24ef13e8749f : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/640dae664dfb4d1a9c9d32e95898761f : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8764139d1b54b058545fd5e0fa98236 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36790a94603249bbb72cb7b79d75e702 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427f2c51b4fa426bbb3f7cfde02b0f95 : SUCCESS in 6m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c7173d395e844c9888840de755358c51 : SUCCESS in 2m 52s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/c5f641b76c2d4b03bf75d998f61fb57e : SUCCESS in 3m 14s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/dc37739bbc27445aa3082b4847bb7263 : SUCCESS in 27m 32s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/9f7639ebeb4e49728afc711da515dfb9 : SUCCESS in 47m 44s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/d8e83468de9c4312b83143df3846bbc1 : SUCCESS in 43m 09s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/a2b84715e10d426e8b70f3675b584490 : SUCCESS in 1h 18m 04s\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/085eeadd7b4147a5831aabb067d05559 : SUCCESS in 15m 50s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/7fb3ae712c144642b0f751724df41d45 : SUCCESS in 8m 32s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/6575a5c0e2a24544ac0dc5fcb143bfb0 : FAILURE in 2h 29m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"029d5d96dacf65d994d5039d8ad4fa208e429340","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-04-07 21:10:55.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"e7607b952018a2271991d01bc04451989ff4aece","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-04-07 21:11:14.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cea9c790b2a0ec55e8e264b77fcb4270feb37f3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 21:22:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1858ce33ce6448a4accaa15b06ff3c25\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0dc056b656e34435aefaca2f79a1cf9c : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee7e45fda5b14a24a2af743517a90145 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ef075e4f6da24b179d38859fabafd41e : SUCCESS in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65da6d21e8e57bf082f8b219a5baa87637ff09fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 23:13:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92948ddd0e7c4ddca459c7d0c0af983c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/953f686d94ae43dca278a7d68add2657 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f03fde7dd894a2996c0e0369e943287 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/153e0d75f4f64c669d3ad3a8140d3127 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a2868158ec0409395c5a5870ad9b373 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2d7b8977ed046b3b114884b6f7b04a2 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee3fc887140e4069a8cc7abd487e213b : SUCCESS in 8m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5f16331704094a439d0f5c4c8d3a446f : SUCCESS in 1m 52s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/5bc47f57d62248c9b1109dc2d229e7d5 : SUCCESS in 3m 33s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/20659a02672841d3948b8cd6b1081c71 : SUCCESS in 30m 12s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/277a34d855f14cefaad747305956d455 : SUCCESS in 1h 15m 04s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/4913743bdfa346288cda720707a0b7a2 : SUCCESS in 1h 08m 23s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/dafd0f43d64740209da394197ffd75c0 : SUCCESS in 46m 26s\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/110e0fb190934deab597c2d19acb0289 : SUCCESS in 16m 25s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/a3f2a4b0be6d44d5a39e2ec13f826603 : SUCCESS in 8m 38s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/9133778f4221480a851ccdc392634628 : FAILURE in 2h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6a00932f3ba0487c0f6dc02c36a6fdb71ea4b22f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-04-11 15:18:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d2084f4181f2651232499893dd40cfa76aac45ed","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-16 08:50:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"38dabd47f937e7326f7177a30a14513162f816de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 08:51:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2489707359130803de7f501b003c9d570a2080ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 10:40:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/488ed75f90ba483ea35146b5e60f0da9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76df1b8de98d4546b1d40c061eb58fad : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc11e2b2342242d6a1dff2f40e89bd2f : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/823d947e7032427ab49f2fb9d95d9380 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d072e2bad074f6690eeabaee276972e : SUCCESS in 8m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3ecf5f707f5945339f5d494bd5803fac : SUCCESS in 4m 53s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/8694026c3a3543828b8fb8fd0c3776ca : SUCCESS in 3m 14s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/d510ac173278407da0f808dd26f4dbfd : SUCCESS in 1h 25m 14s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/17a0f0baa6b24c998f4d4a91d0623aaf : SUCCESS in 1h 29m 49s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/6c33628512f341028ceb66fbfdb701b1 : FAILURE in 1h 20m 31s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/b5c1aa64e8004452b99e60b07290133d : SUCCESS in 1h 09m 59s","accounts_in_message":[],"_revision_number":2},{"id":"4874dc392813eee0339174bf7ecafd9f5ae4fb3b","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-04-17 20:02:35.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f47719b628ff4fa44c7371cc12c58c9e56eb1c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 20:13:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2d997f18d4e475da3323e5bc2de7ba9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b86879d61a8740cd9f074b52d362d485 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5b72292d86a44536919ecce958b8c5fa : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8e018ddb48cf44988e644d35130c1d39 : SUCCESS in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8cded6c69f6538a75a99ea9ab463721ae26cc6d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 23:19:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb73985fdb4d449d9c36204f5e6afc56\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2451e10a8d60401badcbafbc8b11a793 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce49787578bd4dfca053dbbb345aab01 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe6ebb70d99a457f9bf3e684d1e1e1da : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a004b0de849407793ae1aea85bcaff8 : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2518b8c0dab443c3af05ccae2d370790 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72a7a53327748dbb5b7a94bc9cbaaae : SUCCESS in 8m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/38688c0337d745d3928a1096c88e4023 : SUCCESS in 3m 49s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/ae2c591a9fc64a9898c995c070c138e9 : SUCCESS in 3m 01s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/ed306be66c0047aa99f1c2cded70e688 : SUCCESS in 1h 12m 24s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/1071a849833e41f7ae21742d8739df18 : SUCCESS in 1h 08m 12s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/f275c2abfde34b00834dc1b9e4567352 : SUCCESS in 1h 35m 48s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/bffe26d482ca4533b7356044dfb040f3 : SUCCESS in 1h 24m 03s\n- ironic-python-agent-check-image-tinyipa https://zuul.opendev.org/t/openstack/build/d783189bae8340c5858bc6519422016c : SUCCESS in 22m 13s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/62ede9cf3ac84b90b9c411f1ec6df336 : SUCCESS in 8m 29s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/b10134b9de424ccbaf575bcb5ef61f17 : TIMED_OUT in 3h 02m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d01608fabba1a059e54db591b4857ee15a629f07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 23:19:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"84e0a8ca95bc24382f717d2e95c34193d6359853","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 00:53:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd6c25c481ca4524b9c4afc48aa43359\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6f86214a72c4ae0909d19d4674bdc12 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19886352cb0b4946bd5c451106a102ea : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0fdac37082d04eb3b2c7feb453203296 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1c1c7348e7648d0a51835d8b333321b : SUCCESS in 7m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9bb64ca60f9b46e9b84d4b9637f420a3 : SUCCESS in 3m 29s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/b08d2d0c2b584d9b82432af0f83e63a6 : SUCCESS in 2m 45s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/341a98b9d34e49c89d8a3f00ca540c4f : SUCCESS in 46m 27s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/b6f6312118e948bcb87c8a00508aab9c : SUCCESS in 43m 16s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/e1743660feed46d283d9e31f7c03cfc8 : SUCCESS in 42m 06s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/338ec95aceed454dae295083a4e4b085 : SUCCESS in 1h 29m 16s","accounts_in_message":[],"_revision_number":2},{"id":"ecf481f8403b413169c229be09c704f721a5f02d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 00:53:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"857aa77bf6fb8ca22695c04a1b824cb7650e4fbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 00:54:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58e8d11a68484707adb5e10848cbf650\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7456049b68b74f23b30d85563635f769 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a709e6fbb5161bf367215316b1ab11ea62456784","revisions":{"3270ac1450b1710d822356fe98c333a1e5adf416":{"kind":"REWORK","_number":1,"created":"2025-04-07 18:12:40.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/80/946580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/80/946580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/80/946580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/80/946580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/80/946580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/80/946580/1"}}},"commit":{"parents":[{"commit":"7efe3dfc04a69b5f5fc6432e68a13b1c149125c7","subject":"Fix test_timeout on Python3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/7efe3dfc04a69b5f5fc6432e68a13b1c149125c7"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-04-07 18:12:18.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-04-07 18:12:18.000000000","tz":-420},"subject":"Science: base standalone job on redfish standalone","message":"Science: base standalone job on redfish standalone\n\nChange-Id: I2b6d542ce7af3eeeff23f06e43e82de5d7b09701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/3270ac1450b1710d822356fe98c333a1e5adf416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/3270ac1450b1710d822356fe98c333a1e5adf416"}]},"branch":"refs/heads/master"},"a709e6fbb5161bf367215316b1ab11ea62456784":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-04-07 21:11:14.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/80/946580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/80/946580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/80/946580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/80/946580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/80/946580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/80/946580/2"}}},"commit":{"parents":[{"commit":"7efe3dfc04a69b5f5fc6432e68a13b1c149125c7","subject":"Fix test_timeout on Python3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/7efe3dfc04a69b5f5fc6432e68a13b1c149125c7"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-04-07 18:12:18.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-04-07 21:11:14.000000000","tz":0},"subject":"Base standalone job on redfish standalone","message":"Base standalone job on redfish standalone\n\nThe non-redfish standalone job is pending removal from Ironic.\n\nChange-Id: I2b6d542ce7af3eeeff23f06e43e82de5d7b09701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/a709e6fbb5161bf367215316b1ab11ea62456784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/a709e6fbb5161bf367215316b1ab11ea62456784"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
