)]}'
{"id":"openstack%2Fironic-python-agent~954498","triplet_id":"openstack%2Fironic-python-agent~master~Iee7f22e9958b68f263d267c434678478b9a38648","project":"openstack/ironic-python-agent","branch":"master","topic":"valid-partition","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-07-25 07:10:23.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Iee7f22e9958b68f263d267c434678478b9a38648","subject":"Hint at sector sizes when reporting an invalid written image","status":"MERGED","created":"2025-07-09 13:59:34.000000000","updated":"2025-07-25 07:11:46.000000000","submitted":"2025-07-25 07:10:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954498-valid-partition","meta_rev_id":"8ec94cf55595762c0693f776f37f2d7908884449","_number":954498,"virtual_id_number":954498,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-25 07:10:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2025-07-23 08:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2025-07-25 05:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":1,"date":"2025-07-25 05:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-09 14:11:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-09 15:46:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-23 08:30:30.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-07-25 05:31:44.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"}],"messages":[{"id":"0e0b1c4d1065f1a911542a0b4d4996c85ec973cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-07-09 13:59:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2446a51b9530d8734d1270ef0096ca9bac4acaa3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 14:11:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3e10114f899470c922153733d0c1cd2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/56a4b7fa9daa4c70b577a5d3f91d0c0c : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c2a1f531492f4d1bb8871c7c9945cb15 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1f7627d222224e6e880e2ff6b30c6824 : SUCCESS in 9m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3a422f804208758e4d626aae38a5b87a2174e14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 15:46:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f432a33bc294da7880c81c6b572c1e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9df02b2c420d418ea77b1967a0247ca0 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c3c4ab1b4b4a0aa9db4845f3537d24 : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a54521ea2e914933b0108c488e5f4b6e : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ff066572c994a34a71a1eb3611d3c67 : SUCCESS in 2m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01288f0b104447d48477cc55dd0b6b4a : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011ae239d8e1428da65629d297f0b22e : SUCCESS in 4m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5813d58d9e264d369d127907cdd4093a : SUCCESS in 3m 10s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/5315351b04374338907307ab2e52c0d0 : SUCCESS in 1m 55s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/c61cdd52cb3f4c739afdb0a49ad6e1ca : FAILURE in 26m 43s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/775fef7766084cde98f4c88c16de8a4c : FAILURE in 31m 03s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/8ef538776802443cb71a366906ca79d7 : SUCCESS in 1h 41m 12s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/5b794bd3824b414aaac8d3842ee15604 : FAILURE in 58m 05s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/0d04a145e3e54c93b6ffbab39bc16c4f : SUCCESS in 5m 41s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/447aa11d8a344f73bb8928ab67291f96 : SUCCESS in 44m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4285862085b3afcca031e60e63d90d36c0a5eca6","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-07-11 14:20:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e3fc9db7fedf8023abe5f527b867fb35bdf19f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 16:14:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cae4df45da244c2690bba0386658518c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12c2c45f41664046ad4e7c063f7fe03c : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ffdf69617c345db990bc9e214639974 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac19a44506cb4368a9caea8f32a998ec : FAILURE in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7aa54850b86a434c9b8dbba171859038 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e0c9d283fa441a2ab359265128e680f : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f44b16df54d49c280cbbd23cd7f7d82 : SUCCESS in 7m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3c733fc4bfda4a4c91a2ee1ea5f25043 : SUCCESS in 3m 41s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/6f1aac57942c48119bf0764a736d9ad2 : SUCCESS in 3m 19s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/fd6442184211464caf3b5eb36d3f8565 : SUCCESS in 57m 14s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/2f3db1fcdd994e18a8b80d52a8390b4f : SUCCESS in 1h 17m 10s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/2bfdba22e6dd4e6fa5b747f045bbdef5 : SUCCESS in 1h 05m 59s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/3afce97496e44f9598490b08cf973a5e : SUCCESS in 59m 57s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/0aa383b962264a42aa740fcd34eae94a : SUCCESS in 9m 26s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/3a4b3382a39e48a3a8954fef014cc363 : SUCCESS in 1h 48m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31a2056bcd065347317081aa155f6829f10735ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 16:42:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f1ba516491b46b1a46d763291c995d9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cb256aa5f3a14671a4ec2fae2eb874d1 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fbffd2b71cf6458ca97a945115595308 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c20efbf8e8e7470c9d03a1b9823bb83c : SUCCESS in 6m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62d46ca4795c2d517977ed7af663fe6aa30bb3be","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-07-13 14:32:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6ddc91e14be4bbc0d90eb9f1eac071a9044fb4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-13 14:44:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6096e084ded945fbb7ed28324b3b1c86\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/680baff65ed744128613a2a9996a7a54 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c335f5f61169465a93102bdcf42d5e35 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/08364d144c4c42de9656eb00d9b98dbf : SUCCESS in 9m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18ecd9c1a941e44676732f5afd6f71dd848a5d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-13 16:06:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93ad6d63c31b40458fbb8b74fefb269f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92097d8e0d7d4072b5a00afbd832e185 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8675d62bdbd4e2daa53a444a62c88d4 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8d2005d59ec46c69581589e468331fe : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b4072e8955f476eb94ebbde3593e546 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42323dc03e7840128c39b5dc58cc4572 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56712d91e1ba48beb3d40b953f5ad7a8 : SUCCESS in 5m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5d43203389744c08b20e0d5c391301a9 : SUCCESS in 2m 49s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/4a282889b88e48a0a2521290168827a6 : SUCCESS in 4m 35s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/bd116c5d058c46dfa11afcd0e5485559 : SUCCESS in 58m 36s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/60723a917edd499c958eb83ba1df053a : SUCCESS in 1h 09m 50s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/a4cd0b1e5aed43f4b9909dc99f6cbb02 : SUCCESS in 1h 13m 18s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/7c7d0b680ce941f983038ebf83476376 : SUCCESS in 49m 25s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/fc3520ce9ffa42b2b2acc2043e32d79b : SUCCESS in 7m 01s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/06204cf8fefa4029a55f793e6528fd97 : SUCCESS in 1h 31m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6190c956b83e23978683c03219b97a0947d0f886","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-07-21 15:44:28.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"0c7bef206277e1d50772dedba7dc1a6e92bf8138","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-23 08:30:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e96a08967348a78db406ecc0be32924f97f51b02","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2025-07-25 05:31:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1d4ecfc578f506da97383b85f9ad6cbe715c81b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 05:32:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9207dbb041efe70cd830c82f588b7285f2a612e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 07:10:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81c923c1765b49c99baaa0bae5c8b968\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8fc491437f44710987979b188bf6d9d : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fedcd694ada7465999e42d87ebb37c13 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c906d27966b4342ad3c70b902140b50 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/155b2a4c029442399ac0570aa28081e5 : SUCCESS in 6m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/94d2e93d67f645fdb7592d5a942dcec0 : SUCCESS in 4m 07s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/e4c9be8259754edd886fd49e20b635ad : SUCCESS in 2m 51s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/5b6285b8859744848e4a687a5ff0133b : SUCCESS in 1h 02m 27s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/162e821d98a94fb694997a1bbd0a27bb : SUCCESS in 1h 07m 57s\n- metalsmith-integration-ipa-src-uefi https://zuul.opendev.org/t/openstack/build/1eb817a967a74cf2a437c7448ceec9ea : SUCCESS in 1h 33m 19s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/e5a1e917d0e54f5f9670b4b1a96f8193 : SUCCESS in 51m 56s","accounts_in_message":[],"_revision_number":1},{"id":"43d91a0572589a07ad0754fdbdbe0335959a048b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 07:10:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8ec94cf55595762c0693f776f37f2d7908884449","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 07:11:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c13115504ff04beebfd8c6405548c629\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa19cb248bdc433c8d0627da3d2723c7 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e9334fa009525beced643f7ea658f51cf4c3bc4","revisions":{"9e9334fa009525beced643f7ea658f51cf4c3bc4":{"kind":"REWORK","_number":1,"created":"2025-07-09 13:59:34.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/98/954498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/98/954498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/98/954498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/98/954498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/98/954498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/98/954498/1"}}},"commit":{"parents":[{"commit":"883e3cf057aa1e3f7ed5a6fd485e7ab62a8cfd55","subject":"Remove unused tinyipa CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/883e3cf057aa1e3f7ed5a6fd485e7ab62a8cfd55"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-07-09 13:59:03.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-07-09 13:59:03.000000000","tz":120},"subject":"Hint at sector sizes when reporting an invalid written image","message":"Hint at sector sizes when reporting an invalid written image\n\nChange-Id: Iee7f22e9958b68f263d267c434678478b9a38648\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/9e9334fa009525beced643f7ea658f51cf4c3bc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/9e9334fa009525beced643f7ea658f51cf4c3bc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
