)]}'
{"id":"openstack%2Fironic-python-agent~963670","triplet_id":"openstack%2Fironic-python-agent~master~I0adca5ad00ba419a7e2aa6883b3690b4507c25e5","project":"openstack/ironic-python-agent","branch":"master","topic":"ironic-ip-reachable","attention_set":{},"removed_from_attention_set":{"34564":{"account":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"last_update":"2025-10-27 00:08:10.000000000","reason":"\u003cGERRIT_ACCOUNT_34564\u003e replied on the change","reason_account":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-11-10 21:47:01.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-11-10 22:40:49.000000000","reason":"Change was submitted"},"35099":{"account":{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},"last_update":"2025-11-10 22:40:49.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-11-10 15:25:18.000000000","reason":"\u003cGERRIT_ACCOUNT_10239\u003e replied on the change","reason_account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}},"hashtags":["ironic-week-prio"],"change_id":"I0adca5ad00ba419a7e2aa6883b3690b4507c25e5","subject":"Test advertised ip reachability before assigning it","status":"MERGED","created":"2025-10-10 09:38:56.000000000","updated":"2025-11-10 22:42:37.000000000","submitted":"2025-11-10 22:40:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963670-ironic-ip-reachable","meta_rev_id":"b4ee6df2b3363c7440e3f3733d6f080a73801da0","_number":963670,"virtual_id_number":963670,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-10 22:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2025-11-10 21:47:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-10 15:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2025-11-10 21:47:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"}],"CC":[{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-10 10:07:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-10 12:42:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-16 04:35:08.000000000","updated_by":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"reviewer":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"state":"CC"},{"updated":"2025-11-05 22:46:43.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-11-06 08:51:26.000000000","updated_by":{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},"reviewer":{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},"state":"REVIEWER"},{"updated":"2025-11-10 14:58:11.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"853322c4676a572468a27242ffba00f4fe77825c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-10 09:38:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5808cfeb1428c0aa858e83e0d5453ca22967f336","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-10 09:39:12.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"aa585cf939d716e11cbd4b71fcdeb0c4c0ab1967","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-10 09:42:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c8ebe8ef2c59e86ce0d91dee867b1b44d228d60e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 10:07:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00a4311f7c424212ba757b7c2831cda1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/837ccacade0e486d806f4bda3cd6debb : FAILURE in 8m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e3d904c0f3145f684be2788c1588529 : FAILURE in 9m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9bd4af369c074b66bffd3263ac3a24ae : FAILURE in 8m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7392d464555921cae5df4204d6fb859270ed85ba","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-10 10:21:50.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd5f850127811a047aeb64558a59336f5d8c1fc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-10 10:30:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d53b804edad55d14da35a2a7fd6a6414fba939d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 10:59:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd4d10e3ad2d41348e23826dc267dc5e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/239417ddee204254a01704857a14eb3f : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f34d7982ce5c49ebbf44cb3a0d879d68 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb4bf72f3b9f470fa6cf609e406903e6 : SUCCESS in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e312b75cd5beb1f20ad34c89974196c036a3c32e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 12:42:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88ad5df544be4b7d92cc61bd853835c1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/577bd90d27aa494294f5555d3963c415 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21941d603944897820343e1424cd838 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f829e52568ef4c6ebea46e32ef00c0ef : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16e12ec9a8b347b6860e6cbc7c2fbc7f : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae5538bdeb374c3eb0d6466522ff8dfb : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d55cdc412c29419794c6980d595272e3 : SUCCESS in 7m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/15fc0c1027d54c479bca47ac39e21eec : SUCCESS in 2m 37s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/8d7eb476375c49d7bb2e1572eabe9358 : SUCCESS in 4m 48s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/3f69f8e29eb44b2b86df4abd5cb705f6 : FAILURE in 58m 27s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/18f9c5c0ac9148849ed5bc3d1825de4d : FAILURE in 1h 08m 06s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/d199063bbffd49fa8f7b27539217376d : FAILURE in 2h 05m 04s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/671b9ee73a65406ea9adbd8a4258c0fc : SUCCESS in 4m 57s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/481779ddcd7140aa8a00dcb93b780a6e : FAILURE in 49m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4ffe239e0044b4d11dc54323aff0259eb259407","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-10 13:01:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"44f0ea012f111524be50c4dda554874758cbadc2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-10 13:19:25.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82a64fd79ab072bf83110c3feea459473e236aa3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 13:24:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b4902a974344a34a8c79ddbc0dc4d82\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3dc9edc0116248d5a7b0a43d714cbfbf : NODE_FAILURE Node(set) request 35b7721b347d436282cc524d498d5b66 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad31bf35cac84e4aa41dd17ccc88417c : NODE_FAILURE Node(set) request d44ea88c65a74bcb8d2efa57a510db05 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f40484d9131b433d8adb761c18f0da05 : NODE_FAILURE Node(set) request 20b30fd69ea24a50b6a8d9aa9f93e9a9 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cffb7f75f9a96a9f17ffab2842322e1bc5a3de92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-10 13:37:31.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"cc5a4c7308b25f48c950dbbf85847a3582d72f3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-10 13:41:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"73eaa8a219275b714b02612ff7faebd83f7ae982","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 14:33:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a24df71264bd4e919b5f8104d624a9b3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/19a1544487ea4184abf97bf20b853124 : NODE_FAILURE Node(set) request f2f5b84e26b640eb8eeaf3dc3f5905ae failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b62a024a52c94f9693814df36e991678 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/26cf0c8af5c04c53a18577d269bf475f : SUCCESS in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a691c726e4bb126b937b68838e8389e821212802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 14:58:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bac06acbe4c34c4dbeb7099fb85c8777\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8957740a90414ad9a554d0261eed0a0b : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c00abe0d8444fdbb78411912151f12e : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5e0f334819648089e8f269c5e8dd00c : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5683ea1d9b7d418b968c45c52beb394e : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97d96a47e5734040af0eb2c8ee1ea635 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f6c30297a3746818d7147d008b8edf8 : SUCCESS in 8m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c0b03e93372e48fb8b8806794cfaa4c0 : SUCCESS in 2m 52s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/2f463ea5380d4c26ab59b1f4bd6cffa1 : SUCCESS in 3m 30s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/7e7e9882048b4ab59c647b6f19374337 : SUCCESS in 1h 09m 14s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/c8eaa79d8bf14241abe5468f75945db5 : SUCCESS in 1h 09m 05s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/e4e557d1f4024e3ca660f2491bf8cd83 : SUCCESS in 58m 05s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/f976fde5e1c74a809ce1d345eca43edb : SUCCESS in 9m 42s (non-voting)\n- ipa-tempest-ironic-inspector-src https://zuul.opendev.org/t/openstack/build/6a6a6159627741548020dbca3b12d7e7 : SUCCESS in 43m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"40dcfb6f08f8da72fb0d44bc88d17ec05b60b252","author":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"date":"2025-10-16 04:35:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"02a922ca53e01ef440533b5b6c461baea074ead8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-16 13:31:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0d6a414b55726972d01ca137d0cf612e1a075a5c","author":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"date":"2025-10-27 00:08:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"30d0cc679303f95fd1f887382f895cb6b117ff5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-04 09:59:44.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"adebb106c6e364c95a93732a3556ca5ef93f0320","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 10:29:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64e6267e240a4f8c8ba37c46810312cb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ee29e8bca8f4cccbcb5df33fbf56586 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/60af3a3fd4854606a2e220cfe46687b8 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/771eb4050f3c47eba497f363cd0b8b74 : SUCCESS in 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"eea4428ac4932174e78d39f3363788a84efae1df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 11:09:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed78dfcd26584649a54fb09271b0382d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49541a5cfc844b239cac55abc85b2571 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8df3b1ccab490f98e011288f41827a : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0483609c047f46cf93cb440165b4f7ae : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2cfe631b70c47e0b8c7e30878ec3d12 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/127f453182f74dd99c9a665a2d948775 : SUCCESS in 7m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f09db16ca092462b9932cb119ab3866b : SUCCESS in 3m 34s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/d6715b7de6de4637b08383629d6e6f40 : SUCCESS in 4m 41s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/ba287614cacf4241aa3535ff2434391c : SUCCESS in 1h 07m 34s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/223ec1a2d8234cfda2558dca9863be3a : SUCCESS in 47m 53s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/ba2d059f46cb48828a3415b9444dbb4f : SUCCESS in 1h 01m 57s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/558263ccb28f4771bae891016c324b07 : SUCCESS in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3937a259e522d1cfdbf690166ef6fd5dc4780d5f","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-11-05 22:46:43.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ac52ed4b163824429bee07b6b99d5da16021889c","author":{"_account_id":35099,"name":"Jakub Jelinek","email":"kuba.jelinek@nscale.com","username":"kubajj"},"date":"2025-11-06 08:51:26.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"3af88f981e170033f20848875e7ec445672722a0","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-06 15:24:18.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ec6295acec9689504e863e4eb7ea1ed7973b68d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-06 15:28:21.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"c10487bf75bb0230dcec0115bdcbbf9485eea569","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-11-06 15:36:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8b31d7a60b6a865d8d3dad1905f2148924dc28e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 15:45:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aaed54ed317478e9e213975e026dbf9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38c8effcfa1b41aab111ddff6684cb1f : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2f644b842d9499589faad6e99189d3c : NODE_FAILURE Node(set) request a28191fcbc32482599e29f8a89a03e85 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/599482894b3848e1ace1e6b851d350e7 : NODE_FAILURE Node(set) request 26c052418fcc409896b9bcc35dfbf8d9 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a2a111f7457e11f639583b71da33a4c3e1f9954b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 16:47:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/085e45d6ea9b42e78e5c2a536ba2f8ba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c61bcdd4b3414c8395ae84ff5cf468fb : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dabd3414cb04d90a988e65e1efd75e7 : SUCCESS in 6m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1dc18e5f3b8644bfad8fd8ecd5d2ef92 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18f8a4e8acf340c4a5b49e3c1c985dbe : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d235a785ec6475594e79fbd9b38b2e1 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75a0b0e021ea48aeaec83a6fc86872cc : SUCCESS in 3m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2579516614704f3d98910b87be6ad973 : SUCCESS in 3m 57s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/4f3184f6a2ec427ca79c36190d00fa9e : SUCCESS in 3m 37s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/27d1bfa7d6474791a2e79c196797a97b : SUCCESS in 1h 13m 48s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/47f35db29976420981b9caf03abc834d : SUCCESS in 1h 06m 50s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/7c1275b49b974e35a21576e7e2318ed2 : SUCCESS in 1h 04m 11s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/94cca6503248438c97338770d900f9cc : SUCCESS in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6ab006d479b26cb2f6bcf1fd7410aff3ca41db04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-06 17:00:54.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"019d23c17541e80fae089f0d61dcd63e8c491ed6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-06 17:01:32.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0b5a77597b24567e5030876f2e033ba00cfe74ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 17:14:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b094cfa0431404a9019e65c5e93ecc6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03ee213d19d9486a8a0b4c28141995bf : FAILURE in 6m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3acacb51cebc4def8b7aceafc96ec37b : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9fb18bb37fa54a6bb694fd193555bf86 : SUCCESS in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"59ddb5b85c5ec38e23e7e0d5512dfefeb8f7a803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 18:46:02.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2778a9f578348d78eed984c01d8fc6b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e95cbd0389b473ea77560b5ec04a6fd : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac27f91ef8464b1e9a0efd622d584494 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e35d2907cc845f89778c83d5a41d4c0 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fd35379b66c4508bab40dcd0feb8faa : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10aea81a85d246ccbd8f334d5fd93b4b : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/485a8bbe312b4bae88e025e3a90a11e6 : SUCCESS in 6m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ff3e02f39eda45559997ac6549d8105f : SUCCESS in 3m 53s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/2f5551afcc72468faba30cc418d7525d : SUCCESS in 4m 03s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/b9a41cd2287a476db9e8f5b21ecbcdf3 : SUCCESS in 1h 07m 26s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/f556f8706314466fb84e5ae325f1e2d4 : SUCCESS in 1h 07m 03s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/5786f432c1f94de48ec6a4d831dfb4e3 : NODE_FAILURE Node(set) request e865c980ae704fcc93b73418a1134187 failed in 0s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/8d443f1b27c14549ab73fd6e18d513f0 : SUCCESS in 8m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"22d95531323e74e8a40b5d3111efea019657e485","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-07 10:03:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a5ac4e1a6567c8e6a25160a2272903ff873076cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 10:16:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da7507195db74bb8aab70c7f5f954221\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4329c9f4541040efa4ce5f430621549a : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/44d7dbe5694949629bff298d85638187 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/202b45245a254b15aa0e67b3b8fe322f : SUCCESS in 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0d556bdbb185ebbfbfee84edc9bb892721bb8c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 11:32:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/704f5f9bc27e491190de7b5461170b4a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b72470d477c641d2a60b96c62fe55779 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ec7c055ff4049129ba4755c604cc1be : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ecbd4c83e8b4b70b7bb257b49d95681 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/458558790bc243b7a9725cb5c066971c : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed8337c0907f43768c012610fa590217 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/821f2d8d96fc40e99096e671b33448f6 : SUCCESS in 3m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3f27357971d44c36b3e122251daf7754 : SUCCESS in 3m 53s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/abba9e0e17d94f069f0c42a5b1d4a4ea : SUCCESS in 3m 25s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/6e2025af04bf40e3b96b242c7c871c8d : SUCCESS in 1h 18m 03s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/09de360301604af1940fcba534788e9c : SUCCESS in 1h 23m 05s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/dbc1497a0d094a4ea6c61fdc3d4a4f65 : SUCCESS in 57m 56s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/6d32565a441945319cca1fa00953286d : SUCCESS in 9m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"70e8b27d2a1b09b576264c33f095897712b9ef15","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-11-10 14:58:11.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"125ee57dcef15a084738c7415e0bf2bf51aaf88c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-10 15:17:48.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"c2af47c93b39fdd648e021db67e55ddfe53dfaef","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-10 15:18:47.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"114573ab83b905e1819a4c43739ef5c155ab6024","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-11-10 15:25:18.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"4393405683806b41538891a6a529810bba17935f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 15:27:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58d08c5b3ab740be9af56410c9a28d8a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c994d14238174ab1bbad46a8cf260f27 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/efbeab9fd7754685bb68b3193cda81f1 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/194d258249604288a068540e7fcae2d9 : SUCCESS in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"69249b3f6b4b72d321a03714c116629b8a99442a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 16:32:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e7d064b3d4447e98e2c4e9ea37a314a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7463fbb9db7647658c19704cf838c01a : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26a36f83591743a199e2dd239187bdb7 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3322ce3b0d2144ae957018449497a6d4 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/105031454fe74ff8adfad640c02d7bad : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86770f4167a44d7ab332676919d7f3b7 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08e1520686d34c859458c0b4d85dde5a : SUCCESS in 2m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/29d77f061b07427ca65bd97cb26c974c : SUCCESS in 3m 49s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/af8355a05d89403397194ffbb27ca4ad : SUCCESS in 3m 30s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/03982e834a6b41a7bbcfbbbae5030a23 : SUCCESS in 1h 09m 59s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/463357fdb2f448bc8c2dc6fa8f80e841 : SUCCESS in 53m 37s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/2bff041dc4754cb484a3c70838ec3244 : SUCCESS in 47m 44s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/af9a88ce03ff4b12ba50d3d946246c4d : SUCCESS in 9m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4a94501cad5ef19636e80ff72e1615ffc945a1e5","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-11-10 21:47:01.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"9ee61e8b2bf02b01e814a3c5720ea43ccfe89a53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 21:47:26.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"0b58b3d0d89154056985251e256b2fbb442289e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 22:40:49.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81060515170a4e7c93a3ae43fbfdae60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd2b755d50e4d2e9ec7047323f152cc : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcafac0b6f7b4d208680c636414ee344 : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d7d3b08fbea4fa58a57bfa9d02efc7b : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7a9bc50425346cd99d4af5c39032de5 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/170d8415738846ddabfff9724798df72 : SUCCESS in 2m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c8ad957ad24c44f9970b8430ac7c4405 : SUCCESS in 3m 25s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/ddc7b56b30624ce19ba2cd96b77d59df : SUCCESS in 3m 20s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/09da282ab1a44becb24f8e4537531889 : SUCCESS in 48m 24s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/51fedaf9a2f0479ea6f45276a129e001 : SUCCESS in 46m 44s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/4cb23959e6a64ffda45b5a5fc43f5834 : SUCCESS in 49m 27s","accounts_in_message":[],"_revision_number":9},{"id":"2118d9fa97f8a8acb4c5ade1fb8d8c384b89a47a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 22:40:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"b4ee6df2b3363c7440e3f3733d6f080a73801da0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 22:42:37.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3be255e5d1f64974ab8c2933f5f66abd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ac6f5c5374431cbdf74a9c7e867628 : SUCCESS in 1m 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c783a8a9041a46958ee7f2d94441a6cb : SUCCESS in 34s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"2c6cf7cf1f181d60a8abac1b6e910c516dfa73b6","revisions":{"ab661c7b71e7cd05c7da0555fbf65bf793c697d6":{"kind":"REWORK","_number":1,"created":"2025-10-10 09:38:56.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/70/963670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/70/963670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/1"}}},"commit":{"parents":[{"commit":"82fef2db312846fe68e8f7498f11525bd6684c25","subject":"fix: docs(troubleshooting) update deprecated password hash algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/82fef2db312846fe68e8f7498f11525bd6684c25"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 09:30:53.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 09:38:47.000000000","tz":120},"subject":"Test advirtised ip reachability before assigning it","message":"Test advirtised ip reachability before assigning it\n\nThe advertised ip for ironic API is checked only as routable but\nit could still be unreachable, we need to check the actual\nconnectivity before assigning it.\n\nAssisted-By: Claude Sonnet 4\nChange-Id: I0adca5ad00ba419a7e2aa6883b3690b4507c25e5\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/ab661c7b71e7cd05c7da0555fbf65bf793c697d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/ab661c7b71e7cd05c7da0555fbf65bf793c697d6"}]},"branch":"refs/heads/master"},"7c15144931756324f10ff0355597670e8efe0552":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-10-10 09:42:07.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/70/963670/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/70/963670/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/2"}}},"commit":{"parents":[{"commit":"82fef2db312846fe68e8f7498f11525bd6684c25","subject":"fix: docs(troubleshooting) update deprecated password hash algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/82fef2db312846fe68e8f7498f11525bd6684c25"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 09:30:53.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 09:41:52.000000000","tz":120},"subject":"Test advertised ip reachability before assigning it","message":"Test advertised ip reachability before assigning it\n\nThe advertised ip for ironic API is checked only as routable but\nit could still be unreachable, we need to check the actual\nconnectivity before assigning it.\n\nAssisted-By: Claude Sonnet 4\nChange-Id: I0adca5ad00ba419a7e2aa6883b3690b4507c25e5\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/7c15144931756324f10ff0355597670e8efe0552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/7c15144931756324f10ff0355597670e8efe0552"}]},"branch":"refs/heads/master"},"e0e2b3e857d00df9be1f3cd1a24b98ec1522fca9":{"kind":"REWORK","_number":3,"created":"2025-10-10 10:30:54.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/70/963670/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/70/963670/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/3"}}},"commit":{"parents":[{"commit":"82fef2db312846fe68e8f7498f11525bd6684c25","subject":"fix: docs(troubleshooting) update deprecated password hash algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/82fef2db312846fe68e8f7498f11525bd6684c25"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 09:30:53.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 10:30:47.000000000","tz":120},"subject":"Test advertised ip reachability before assigning it","message":"Test advertised ip reachability before assigning it\n\nThe advertised ip for ironic API is checked only as routable but\nit could still be unreachable, we need to check the actual\nconnectivity before assigning it.\n\nAssisted-By: Claude Sonnet 4\nChange-Id: I0adca5ad00ba419a7e2aa6883b3690b4507c25e5\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/e0e2b3e857d00df9be1f3cd1a24b98ec1522fca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/e0e2b3e857d00df9be1f3cd1a24b98ec1522fca9"}]},"branch":"refs/heads/master"},"0b7e54fc173f2f8950578026caa5f023df80138b":{"kind":"REWORK","_number":4,"created":"2025-10-10 13:37:31.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/70/963670/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/70/963670/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/4"}}},"commit":{"parents":[{"commit":"82fef2db312846fe68e8f7498f11525bd6684c25","subject":"fix: docs(troubleshooting) update deprecated password hash algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/82fef2db312846fe68e8f7498f11525bd6684c25"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 09:30:53.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 13:37:26.000000000","tz":120},"subject":"Test advertised ip reachability before assigning it","message":"Test advertised ip reachability before assigning it\n\nThe advertised ip for ironic API is checked only as routable but\nit could still be unreachable, we need to check the actual\nconnectivity before assigning it.\n\nAssisted-By: Claude Sonnet 4\nChange-Id: I0adca5ad00ba419a7e2aa6883b3690b4507c25e5\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/0b7e54fc173f2f8950578026caa5f023df80138b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/0b7e54fc173f2f8950578026caa5f023df80138b"}]},"branch":"refs/heads/master"},"66ac8a291b4a5dc6b5d3735778e3ae17b24ac60e":{"kind":"REWORK","_number":5,"created":"2025-10-10 13:41:50.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/70/963670/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/70/963670/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/5"}}},"commit":{"parents":[{"commit":"82fef2db312846fe68e8f7498f11525bd6684c25","subject":"fix: docs(troubleshooting) update deprecated password hash algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/82fef2db312846fe68e8f7498f11525bd6684c25"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 09:30:53.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 13:41:45.000000000","tz":120},"subject":"Test advertised ip reachability before assigning it","message":"Test advertised ip reachability before assigning it\n\nThe advertised ip for ironic API is checked only as routable but\nit could still be unreachable, we need to check the actual\nconnectivity before assigning it.\n\nAssisted-By: Claude Sonnet 4\nChange-Id: I0adca5ad00ba419a7e2aa6883b3690b4507c25e5\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/66ac8a291b4a5dc6b5d3735778e3ae17b24ac60e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/66ac8a291b4a5dc6b5d3735778e3ae17b24ac60e"}]},"branch":"refs/heads/master"},"9dc563cf575fca994d08516206f2e9ee6c18fecd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-11-04 09:59:44.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/70/963670/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/70/963670/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/6"}}},"commit":{"parents":[{"commit":"6fdb13d6d40d6f3b786dba8204579dca1d434341","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/6fdb13d6d40d6f3b786dba8204579dca1d434341"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 09:30:53.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-04 09:59:36.000000000","tz":60},"subject":"Test advertised ip reachability before assigning it","message":"Test advertised ip reachability before assigning it\n\nThe advertised ip for ironic API is checked only as routable but\nit could still be unreachable, we need to check the actual\nconnectivity before assigning it.\n\nAssisted-By: Claude Sonnet 4\nChange-Id: I0adca5ad00ba419a7e2aa6883b3690b4507c25e5\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/9dc563cf575fca994d08516206f2e9ee6c18fecd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/9dc563cf575fca994d08516206f2e9ee6c18fecd"}]},"branch":"refs/heads/master"},"e35ee87c6965b5426fbda9cf7306573d7f88e9c4":{"kind":"REWORK","_number":7,"created":"2025-11-06 15:28:21.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/70/963670/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/70/963670/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/7"}}},"commit":{"parents":[{"commit":"bb4b4fdb3887c1fbe0e8536869ac68e9ed1dbe96","subject":"Fix for matching hints with lists of strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/bb4b4fdb3887c1fbe0e8536869ac68e9ed1dbe96"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 09:30:53.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-06 15:28:14.000000000","tz":60},"subject":"Test advertised ip reachability before assigning it","message":"Test advertised ip reachability before assigning it\n\nThe advertised ip for ironic API is checked only as routable but\nit could still be unreachable, we need to check the actual\nconnectivity before assigning it.\n\nAssisted-By: Claude Sonnet 4\nChange-Id: I0adca5ad00ba419a7e2aa6883b3690b4507c25e5\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/e35ee87c6965b5426fbda9cf7306573d7f88e9c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/e35ee87c6965b5426fbda9cf7306573d7f88e9c4"}]},"branch":"refs/heads/master"},"1dcf742c77df1323d418f9c2bdab8807fc4fdd5b":{"kind":"REWORK","_number":8,"created":"2025-11-06 17:00:54.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/70/963670/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/70/963670/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/8"}}},"commit":{"parents":[{"commit":"bb4b4fdb3887c1fbe0e8536869ac68e9ed1dbe96","subject":"Fix for matching hints with lists of strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/bb4b4fdb3887c1fbe0e8536869ac68e9ed1dbe96"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 09:30:53.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-06 17:00:41.000000000","tz":60},"subject":"Test advertised ip reachability before assigning it","message":"Test advertised ip reachability before assigning it\n\nThe advertised ip for ironic API is checked only as routable but\nit could still be unreachable, we need to check the actual\nconnectivity before assigning it.\n\nAssisted-By: Claude Sonnet 4\nChange-Id: I0adca5ad00ba419a7e2aa6883b3690b4507c25e5\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/1dcf742c77df1323d418f9c2bdab8807fc4fdd5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/1dcf742c77df1323d418f9c2bdab8807fc4fdd5b"}]},"branch":"refs/heads/master"},"2c6cf7cf1f181d60a8abac1b6e910c516dfa73b6":{"kind":"REWORK","_number":9,"created":"2025-11-10 15:17:48.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/70/963670/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/70/963670/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/70/963670/9"}}},"commit":{"parents":[{"commit":"bb4b4fdb3887c1fbe0e8536869ac68e9ed1dbe96","subject":"Fix for matching hints with lists of strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/bb4b4fdb3887c1fbe0e8536869ac68e9ed1dbe96"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-10-10 09:30:53.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-10 15:17:30.000000000","tz":60},"subject":"Test advertised ip reachability before assigning it","message":"Test advertised ip reachability before assigning it\n\nThe advertised ip for ironic API is checked only as routable but\nit could still be unreachable, we need to check the actual\nconnectivity before assigning it.\n\nAssisted-By: Claude Sonnet 4\nChange-Id: I0adca5ad00ba419a7e2aa6883b3690b4507c25e5\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/2c6cf7cf1f181d60a8abac1b6e910c516dfa73b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/2c6cf7cf1f181d60a8abac1b6e910c516dfa73b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
