)]}'
{"id":"openstack%2Fironic-python-agent~994982","triplet_id":"openstack%2Fironic-python-agent~master~I2dce3d8fdeccb9bd8fe63ebad62c3206e30235aa","project":"openstack/ironic-python-agent","branch":"master","topic":"ipa-centos10","attention_set":{},"removed_from_attention_set":{"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2026-09-17 22:39:40.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I2dce3d8fdeccb9bd8fe63ebad62c3206e30235aa","subject":"Switch ironic-standalone-ipa-src to centos10 ipa","status":"MERGED","created":"2026-06-25 21:52:08.000000000","updated":"2026-09-17 22:41:52.000000000","submitted":"2026-09-17 22:39:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"994982-ipa-centos10","meta_rev_id":"f987bab006657f116372d413d0d6ebeb5191de05","_number":994982,"virtual_id_number":994982,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 22:39:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2026-09-17 15:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2026-09-17 15:35:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2026-09-17 15:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2026-09-17 15:32:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-25 22:31:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-26 00:24:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 08:39:08.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2026-09-17 15:00:00.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2026-09-17 15:32:48.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-09-17 15:35:31.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"7ce7ff5e614f158bd6f95780e5c25ad919b3505b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-06-25 21:52:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3964beba74e45305480b3f038cfd7ad7335f508b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 22:31:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/276b7eab485745a19c53a2a4d226cd8f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/de65c91a4d644170bc49d32b6ea029ef : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d1574031531248988a9f7ba7f34e6bef : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/784273e584d64818a435b604b9e50744 : SUCCESS in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8919c663a711ca2479e32baa75297724d251fafd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 00:24:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b07fc97530324e0891620026a0e0d7ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21f0ff7a7ff14ea1a8fca58bb65aede9 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe27ff777c2d447abe3fa7c19e7bdfe7 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bb60757ced54b79a4ceb1eef83c36c9 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83274a8f156245e4a4f39a8a338f2587 : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a40fe10450f40b29bd52e34491000d7 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76bcd9c537cd4725b27168f3829a47a4 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd5daa4f0da24169af7d12fe85df7be8 : SUCCESS in 5m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/65b422aabf1049aa80dc3e64b4eccd3f : SUCCESS in 3m 58s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/18b1208d0a6e4dab94c309621e8ac50d : SUCCESS in 3m 24s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/92493c55478d4a59b5301f89998fcb1e : SUCCESS in 33m 19s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/625bfccbc2874fd58604ffa15dd10c12 : SUCCESS in 1h 09m 03s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/7098574eb9684f1f8a550eb6f34495db : FAILURE in 2h 29m 42s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/71d23324913f4115af80bd002435bb39 : SUCCESS in 10m 32s (non-voting)\n- ironic-python-agent-check-image-dib-centos10 https://zuul.opendev.org/t/openstack/build/f86006edc9c046a198e55324ba10d5dd : SUCCESS in 8m 23s (non-voting)\n- ironic-python-agent-check-image-dib-debian https://zuul.opendev.org/t/openstack/build/bffb8b31702e40d4972b8205b3a6fd05 : SUCCESS in 4m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a7ca1cc05fb91ac16ca181a78b76a9789aef83c","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-06-30 21:13:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0dce9ebfeefa16498b1b944f8e83ffb9ce6263e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 21:25:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19fb4362d31f4ba79ca738305ce9e535\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f71575b71cb84589ad92470114130292 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1848b8fe967647b896c69aaada70459a : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/667b9654bd144cf38ced1265a4db9796 : SUCCESS in 9m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6df145e0e3ce4c86478d45c4f62852a0b0656613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 22:33:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0534ab59dc62460693813562a30dbe7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2da05e43e4314104a627a0846379c108 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b857e87e0d84e06b17431990ceaf7da : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e7860cbca164d27a8fb91a5163dcdd8 : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3aa097476d484930b970f447e0280753 : SUCCESS in 2m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/84f7750a142f4b3a94f9324f9601598d : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d08d32a0c1496b8b2006c087f8c64a : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85bf137791a040c99bdd540cc9ab864f : SUCCESS in 5m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2446c57addb74e3b93ef25388d582374 : SUCCESS in 5m 14s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/e77bb369f690410e860d30033c04f5b6 : SUCCESS in 3m 12s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/89710c3d264f4034b94905b016bab0ae : FAILURE in 30m 47s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/6dc996ad12d64961b535d51197ade500 : SUCCESS in 1h 14m 05s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/51d9a4f7024142e5aad5b83470ccbeec : SUCCESS in 56m 42s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/0947e6e043504f68bb31afcd259f7cd8 : SUCCESS in 13m 06s (non-voting)\n- ironic-python-agent-check-image-dib-centos10 https://zuul.opendev.org/t/openstack/build/735bf8f82869494fb1a78875fc7f67ef : SUCCESS in 10m 09s (non-voting)\n- ironic-python-agent-check-image-dib-debian https://zuul.opendev.org/t/openstack/build/76b72fd9ca7341e6bd71ec693d627ca6 : SUCCESS in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ed1509372b7a40c55408bc0a79357109316abd1","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-09-16 08:39:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44b89a186925336383f3abd128fce1068e1ee190","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 08:48:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17b4d0ad266c45b5a42044a72fb5d3a6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3cb433844ef84981b13f3f78bc71a475 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f2f1eff65fc2443ca49cdc73020ae246 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3a8aa96f74b34f3dbc1e9f5bcffae22c : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c904b037d681311df1cf6037664108c2f94ad616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 09:54:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a9c184b27ee4cb7a2f7fff39c0ad5dd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6010e10b567249be8a9effb7b5dd7864 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3478c2d3c2474e3b8f2ac1a575867eb6 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78c1af3120b8415f9027de927740bae2 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/070464fd772a4adf89b984b1805bab26 : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1d3a1e121a1c429f96efc80fc282a89e : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/945e452e60774760b437f1f26a0dde14 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/116bab2a9fdb4fe5bfebd411b26305bf : SUCCESS in 5m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1dc511a1991b4f3395917e7b0a207c2c : SUCCESS in 3m 47s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/6335d400f77f476980099e5b31955658 : SUCCESS in 3m 20s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/fecdb7d45d204f3089df7cc66bae2cc2 : SUCCESS in 32m 35s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/d269f7a5eef54795ad7174ead5e45e4e : SUCCESS in 1h 09m 58s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/5a9b3e35349748b9bf11d0f04d24366e : SUCCESS in 50m 38s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/5c8f4f592b6746dea7dadc80368e9f2d : SUCCESS in 12m 42s (non-voting)\n- ironic-python-agent-check-image-dib-centos10 https://zuul.opendev.org/t/openstack/build/161360f7edf5474388b7fb97bbe01679 : SUCCESS in 11m 22s (non-voting)\n- ironic-python-agent-check-image-dib-debian https://zuul.opendev.org/t/openstack/build/7d3b5e57b8d44f32a7e8d7ecc3bff78a : SUCCESS in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf77b4f3a78914774311cf574f83b43bef06c139","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-09-17 15:00:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"663ef6b14b49dd2b303db0fa42d361d9cf27bc8e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-09-17 15:00:11.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"5d283171676ee6337b1e17c773e94d44058a5514","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-09-17 15:32:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9ab20e1ad408625b6bd7709811e44d19974b2cd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 15:33:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4295ee453d11cb8363d01a930c6449d3d61f1563","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-09-17 15:35:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3109db3c733cba85b3c1662144cb7eb4954bfc13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:39:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11b2fd0abed74dfb87c197a5382597c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c1dce2ebe24d788fa4defe27f46dda : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58d2387618454280aa6529fe0c858ecd : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd204e57e9424ad0953a6af9f15e3763 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb6c99bbcd7e4719b8de17ef18440378 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9aee7029f9a47bfaeabd28f95253480 : SUCCESS in 7m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a5863a866f2345528b8f5dc1f8f04455 : SUCCESS in 2m 19s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/9cc2c9c0661d40158aa13759c9aadebd : SUCCESS in 3m 27s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/8b9f3835c5bc498c8afcc4abe3eb9906 : SUCCESS in 1h 05m 48s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/c7c7bb8abbc442a2878a01942b2d2a24 : SUCCESS in 1h 01m 13s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/e5320d3097f144db87a0fdcbcaaf4aa9 : FAILURE in 17m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4d14b9d2fc13d730671e223d81db7c7dfbda102d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-09-17 19:57:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba8e93d7b4313fee61107906e3467070a42f6509","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 20:09:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d50b344c937496fbb0c0f62d1d4a28a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6027cdd112834ea98a7ad52e71533965 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/39da867a088c4e11ac3af9f0b708f32f : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/993e36f6414a40baaf6d9e9a848dd553 : SUCCESS in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf13b95834add1334722f762f550400c38e29212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 21:15:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9debdbbfe7fb4892b095300fe17df99d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/853971ffde174483a7183c6580b6e647 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36ce783b8c57454890d17b522d4b3639 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c708fcfa17c4fa383c0193492573ed4 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6624fd1f7dff453caaa553dd9979a389 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/19f7d9cd08184af9b33a82e904c13039 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c93b9fa7e804a0dade6550755163ba0 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a33c240c3ca4bcb9869f911c9663220 : SUCCESS in 6m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dd201400ec7747f18088fc76849455b5 : SUCCESS in 3m 48s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/25280cd3f7b949ddbd5b7047983fd893 : SUCCESS in 3m 31s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/8ddccc4768b64fa19e29c508174e9033 : SUCCESS in 1h 11m 33s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/40480a21d896426486d9e95c8e9bf489 : SUCCESS in 59m 28s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/70797cc249ad4979943177d46311143d : SUCCESS in 51m 52s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/c9088119264a417080ad1f7fc6bb7383 : SUCCESS in 6m 49s (non-voting)\n- ironic-python-agent-check-image-dib-centos10 https://zuul.opendev.org/t/openstack/build/40b68f9db4594f0b9effc05c0a67fa9f : SUCCESS in 11m 23s (non-voting)\n- ironic-python-agent-check-image-dib-debian https://zuul.opendev.org/t/openstack/build/a16c744928fb48909a71e65b69e78da5 : SUCCESS in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8853e63f121ab19228394043082d0d70e8aaf0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 21:16:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f3bb561fa07a16fda9e7d174b896c53e4011ecf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 22:39:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61d690e3001b4f21a5d59bb15bb3ebec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ab917e3d8fd4605801d81b41d68313d : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f64b2b2293ba478ebae8953f2972ddb7 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8a6d5c62e4243cab3598655410823cf : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/772b0f9275304655ad07b4a80c98ea19 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f2a0cc533e54ad8af1c764957e14e55 : SUCCESS in 3m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/00efd871feeb47fd80f2aca456a4c09b : SUCCESS in 2m 19s\n- ipa-tox-bandit https://zuul.opendev.org/t/openstack/build/9dda02ec355f4cfc9b50ce4f1acf483c : SUCCESS in 2m 18s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/0fec26022c384175ab20d460624fdb02 : SUCCESS in 1h 17m 20s\n- ipa-tempest-uefi-redfish-vmedia-src https://zuul.opendev.org/t/openstack/build/bca01c3666b24e1182efe2fd0fcca12e : SUCCESS in 36m 27s\n- ironic-standalone-ipa-src https://zuul.opendev.org/t/openstack/build/5f9b254da4b349dcb72647cf21526390 : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fc3ac363c884c78bf9025d44513942c8a051694c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 22:39:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f987bab006657f116372d413d0d6ebeb5191de05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 22:41:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fae0c29c33d843f7bdb57dede7f068a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13981136c1dc466bb89a21a10531d76c : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e6f46f81f957e2c5dc3c5815f1b205e98f0507f","revisions":{"6e6f46f81f957e2c5dc3c5815f1b205e98f0507f":{"kind":"REWORK","_number":1,"created":"2026-06-25 21:52:08.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/82/994982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-python-agent","ref":"refs/changes/82/994982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/82/994982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/82/994982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-python-agent refs/changes/82/994982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-python-agent refs/changes/82/994982/1"}}},"commit":{"parents":[{"commit":"80895ff5d2c13a45cf519065e15cfdb01ac13c06","subject":"Enable centos10 publish job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/80895ff5d2c13a45cf519065e15cfdb01ac13c06"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-06-25 21:51:05.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-06-25 21:51:25.000000000","tz":720},"subject":"Switch ironic-standalone-ipa-src to centos10 ipa","message":"Switch ironic-standalone-ipa-src to centos10 ipa\n\nDepends-On: I7801f706566b5e06080c545e62855e5c742900bb\nChange-Id: I42d11b05136d3dc4cd34d9dcfafadd459dfffba1\n\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nChange-Id: I2dce3d8fdeccb9bd8fe63ebad62c3206e30235aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/6e6f46f81f957e2c5dc3c5815f1b205e98f0507f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-python-agent/commit/6e6f46f81f957e2c5dc3c5815f1b205e98f0507f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
