)]}'
{"id":"openstack%2Fironic-specs~658556","triplet_id":"openstack%2Fironic-specs~master~I8f0fe7ac439afb5d98a386317f3df099b1582765","project":"openstack/ironic-specs","branch":"master","hashtags":[],"change_id":"I8f0fe7ac439afb5d98a386317f3df099b1582765","subject":"Rename review.openstack.org to review.opendev.org","status":"ABANDONED","created":"2019-05-11 19:17:21.000000000","updated":"2020-09-23 00:54:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"66272974cc9c1052951c2502e84ab97c429ac0c0","_number":658556,"virtual_id_number":658556,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":-1,"date":"2019-05-13 14:49:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 10:22:51.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-05-13 14:49:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30dbc1c6a45741868b0031a67b1883211f35364c","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-11 19:17:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb1d62bfe5a751ca2ec4263b99e93a1f484f98fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 19:29:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/56/658556/1/check/openstack-tox-docs/6cee085/html/ : SUCCESS in 6m 53s\n- openstack-tox-linters http://logs.openstack.org/56/658556/1/check/openstack-tox-linters/707fbab/ : FAILURE in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"56230c99454b6e7b6ea269f27704ba22889cbb57","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-11 19:56:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"687600c2297a3eaead428c1c21d88f0eea9da9a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 20:08:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/56/658556/2/check/openstack-tox-docs/fc93f9b/html/ : SUCCESS in 6m 57s\n- openstack-tox-linters http://logs.openstack.org/56/658556/2/check/openstack-tox-linters/1aa825f/ : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"5ed550deb0ad49c66d95fdf98c816d2a6ae98723","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-13 10:22:51.000000000","message":"Patch Set 2: Code-Review-1\n\nHi, please run tox locally and verify it passes.\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"6715b8bb30ca725acc0692e3564b5fe91c6e835d","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-13 14:34:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dcedc943f059050fbaa83e128ec47c055c8efad5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 14:49:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/56/658556/3/check/openstack-tox-docs/8d33ba2/html/ : SUCCESS in 7m 53s\n- openstack-tox-linters http://logs.openstack.org/56/658556/3/check/openstack-tox-linters/aeacccd/ : FAILURE in 5m 22s","accounts_in_message":[],"_revision_number":3},{"id":"66272974cc9c1052951c2502e84ab97c429ac0c0","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-09-23 00:54:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7c5f1c60a70f534bd975a57d6e77961d9b82275c","revisions":{"36214859c96b51368449efb53a984d19770eafec":{"kind":"REWORK","_number":1,"created":"2019-05-11 19:17:21.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/56/658556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/56/658556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/1"}}},"commit":{"parents":[{"commit":"6d4c0785e3863bd809c3c84926fc9c888da1fd05","subject":"Merge \"OOB disk-erase for ilo5 based HPE Proliant Servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/6d4c0785e3863bd809c3c84926fc9c888da1fd05"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-11 19:16:30.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-11 19:17:18.000000000","tz":480},"subject":"Rename review.openstack.org to review.opendev.org","message":"Rename review.openstack.org to review.opendev.org\n\nThere are many references to review.openstack.org, and while the\nredirect should work, we can also go ahead and fix them.\n\nChange-Id: I8f0fe7ac439afb5d98a386317f3df099b1582765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/36214859c96b51368449efb53a984d19770eafec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/36214859c96b51368449efb53a984d19770eafec"}]},"branch":"refs/heads/master"},"931d97be8d1e353926b3f74e550429d6e012a056":{"kind":"REWORK","_number":2,"created":"2019-05-11 19:56:03.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/56/658556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/56/658556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/2"}}},"commit":{"parents":[{"commit":"6d4c0785e3863bd809c3c84926fc9c888da1fd05","subject":"Merge \"OOB disk-erase for ilo5 based HPE Proliant Servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/6d4c0785e3863bd809c3c84926fc9c888da1fd05"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-11 19:16:30.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-11 19:55:52.000000000","tz":480},"subject":"Rename review.openstack.org to review.opendev.org","message":"Rename review.openstack.org to review.opendev.org\n\nThere are many references to review.openstack.org, and while the\nredirect should work, we can also go ahead and fix them.\n\nChange-Id: I8f0fe7ac439afb5d98a386317f3df099b1582765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/931d97be8d1e353926b3f74e550429d6e012a056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/931d97be8d1e353926b3f74e550429d6e012a056"}]},"branch":"refs/heads/master"},"7c5f1c60a70f534bd975a57d6e77961d9b82275c":{"kind":"REWORK","_number":3,"created":"2019-05-13 14:34:38.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/56/658556/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/56/658556/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/56/658556/3"}}},"commit":{"parents":[{"commit":"6d4c0785e3863bd809c3c84926fc9c888da1fd05","subject":"Merge \"OOB disk-erase for ilo5 based HPE Proliant Servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/6d4c0785e3863bd809c3c84926fc9c888da1fd05"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-11 19:16:30.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-13 14:34:14.000000000","tz":480},"subject":"Rename review.openstack.org to review.opendev.org","message":"Rename review.openstack.org to review.opendev.org\n\nThere are many references to review.openstack.org, and while the\nredirect should work, we can also go ahead and fix them.\n\nChange-Id: I8f0fe7ac439afb5d98a386317f3df099b1582765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/7c5f1c60a70f534bd975a57d6e77961d9b82275c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/7c5f1c60a70f534bd975a57d6e77961d9b82275c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
