)]}'
{"id":"openstack%2Fironic-specs~718365","triplet_id":"openstack%2Fironic-specs~master~I975a7012c42b8ee2c30d3a33fc63a083738d8007","project":"openstack/ironic-specs","branch":"master","topic":"cleanup-py27-support","hashtags":[],"change_id":"I975a7012c42b8ee2c30d3a33fc63a083738d8007","subject":"Cleanup py27 support","status":"MERGED","created":"2020-04-08 09:51:11.000000000","updated":"2020-04-09 08:58:41.000000000","submitted":"2020-04-09 08:55:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718365-1586422529621-0a348a98","meta_rev_id":"0e24d6fe856108c5825d917212f88ce00220d304","_number":718365,"virtual_id_number":718365,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 08:55:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-09 08:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-09 08:37:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 08:37:25.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-09 08:55:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a65bfb1fa82da32bdf2fed4c9a5fae807ed5ecb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-08 09:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c65691accb67f35cecbbf162177ad376de40d92c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 10:08:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f937113039b64f8ebbc24c2d37be4dd1 : SUCCESS in 11m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5aca5a4db56a424f9ab26329002743d8 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3fbd92ff96c92447d4381029ccfeb8423ed6cf41","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-09 08:37:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d19363f2adfe5db57bb4bdabe92e52d2bf24e219","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 08:37:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b09c7505c2c222dfaa9e470a1679c45cb90b762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 08:55:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0eecfe8a83540ccb08c0070d5aada41 : SUCCESS in 12m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e83aa8196d0d4c9aba2eecd2466bf258 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c6d37b0de841f7acad23b49215e0e0c0a3f7f55a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 08:55:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5b7e4666ba112b88df927b8ae2f69c07dc4a9698","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 08:58:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/aea709cc8d294e2b8d00e780813fdb7c : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3de1d4cec1817b4c4d166d67ed01e6673b27aed9","revisions":{"3de1d4cec1817b4c4d166d67ed01e6673b27aed9":{"kind":"REWORK","_number":1,"created":"2020-04-08 09:51:11.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/65/718365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/65/718365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/65/718365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/65/718365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/65/718365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/65/718365/1"}}},"commit":{"parents":[{"commit":"e589193802f36ba2c066e4d3b6b2c098b24de789","subject":"Merge \"Update L3 based deployment spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/e589193802f36ba2c066e4d3b6b2c098b24de789"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-08 09:49:00.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-08 09:50:56.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove obsolete sections from setup.cfg\n- Update requirements, no need for python_version anymore\n- Require newer pbr\n- Remove install_command, the default is fine\n\nChange-Id: I975a7012c42b8ee2c30d3a33fc63a083738d8007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/3de1d4cec1817b4c4d166d67ed01e6673b27aed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/3de1d4cec1817b4c4d166d67ed01e6673b27aed9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
