)]}'
{"id":"openstack%2Fironic-specs~756231","triplet_id":"openstack%2Fironic-specs~master~Iba10744bc935dcc6837cd673a8fc6a45e8a14263","project":"openstack/ironic-specs","branch":"master","hashtags":[],"change_id":"Iba10744bc935dcc6837cd673a8fc6a45e8a14263","subject":"Remove usage of six","status":"ABANDONED","created":"2020-10-06 08:59:29.000000000","updated":"2020-10-13 00:49:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"78fc33576cdeccc188dd7720f547099980ca816d","_number":756231,"virtual_id_number":756231,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-06 09:11:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 09:11:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32eb2c85a3680eb35ce8d3c340869cd85add8b81","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-10-06 08:59:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce97ad242818fc5216ec779a709dea386ddc9442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 09:11:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d21e72f938419c8edb224564aeb98e : SUCCESS in 9m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/59cf5e5c0bef44f9a120d3472c2c5f39 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"78fc33576cdeccc188dd7720f547099980ca816d","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-10-13 00:49:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a1b0a7977b07ed747fcaf2f0021b0a42025f0b0","revisions":{"4a1b0a7977b07ed747fcaf2f0021b0a42025f0b0":{"kind":"REWORK","_number":1,"created":"2020-10-06 08:59:29.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/31/756231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/31/756231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/31/756231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/31/756231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/31/756231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/31/756231/1"}}},"commit":{"parents":[{"commit":"4c616040db7fd135f348f09bcee59a195a9ca14b","subject":"Merge \"New release model: correct the named branches lifetime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/4c616040db7fd135f348f09bcee59a195a9ca14b"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-06 08:58:08.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-06 08:59:22.000000000","tz":480},"subject":"Remove usage of six","message":"Remove usage of six\n\nChange-Id: Iba10744bc935dcc6837cd673a8fc6a45e8a14263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/4a1b0a7977b07ed747fcaf2f0021b0a42025f0b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/4a1b0a7977b07ed747fcaf2f0021b0a42025f0b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
