)]}'
{"id":"openstack%2Fironic-specs~759839","triplet_id":"openstack%2Fironic-specs~master~I3849e4e644a6e049ca48a734e45e70f35181afbc","project":"openstack/ironic-specs","branch":"master","topic":"pxe-boot-from-volume","hashtags":[],"change_id":"I3849e4e644a6e049ca48a734e45e70f35181afbc","subject":"Add PXE boot from iSCSI volume scenario","status":"ABANDONED","created":"2020-10-27 10:54:23.000000000","updated":"2020-11-12 08:40:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"53837881fc0ea899412bd3b301736fa47b3cd10e","_number":759839,"virtual_id_number":759839,"owner":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-27 14:47:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-10-30 09:38:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-27 14:47:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-30 09:38:06.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"78b713a3cec7fdcd401cbac1b97f78b92b6397a2","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-10-27 10:54:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9456bbc2daa91b55f8f053485ad8b43a8b96583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 11:04:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d20878faac437f88e1cf5afa698cae : SUCCESS in 8m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/02a71da9e2784348881349e80ac3303a : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c109856578d853e6f98dbc85c475d2f4cd8ad453","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-10-27 14:10:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5842ae1859b37680b5bc5e7e2cd85a689957e5de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 14:47:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74fa5efbc2f442409eaddce657a4191d : SUCCESS in 19m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7a88e8b1e3d645d4ab89d3f70fc97e31 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f96dd91fa1d6cca02853329c6184341108962227","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-10-30 09:38:06.000000000","message":"Patch Set 2: Code-Review-1\n\nLet\u0027s probably leave this spec as it is for historical purposes. The new work can be tracked as an RFE.","accounts_in_message":[],"_revision_number":2},{"id":"f8e5ff5c634ac78bc41898a4f20f42745ef75e48","author":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"date":"2020-11-12 08:40:44.000000000","message":"Abandoned\n\nCreate a new spec:\nhttps://review.opendev.org/#/c/762474/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82c4fc46f7c8d4f3c66f668931d1fd3af8887a06","revisions":{"e37e56634ec7e4b26735b5605a12c46e6bf84652":{"kind":"REWORK","_number":1,"created":"2020-10-27 10:54:23.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/39/759839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/39/759839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/39/759839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/39/759839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/39/759839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/39/759839/1"}}},"commit":{"parents":[{"commit":"4c616040db7fd135f348f09bcee59a195a9ca14b","subject":"Merge \"New release model: correct the named branches lifetime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/4c616040db7fd135f348f09bcee59a195a9ca14b"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-10-27 10:54:08.000000000","tz":0},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-10-27 10:54:08.000000000","tz":0},"subject":"Add PXE boot from iSCSI volume scenario","message":"Add PXE boot from iSCSI volume scenario\n\nChange-Id: I3849e4e644a6e049ca48a734e45e70f35181afbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/e37e56634ec7e4b26735b5605a12c46e6bf84652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/e37e56634ec7e4b26735b5605a12c46e6bf84652"}]},"branch":"refs/heads/master"},"82c4fc46f7c8d4f3c66f668931d1fd3af8887a06":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-27 14:10:18.000000000","uploader":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"ref":"refs/changes/39/759839/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/39/759839/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/39/759839/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/39/759839/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/39/759839/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/39/759839/2"}}},"commit":{"parents":[{"commit":"4c616040db7fd135f348f09bcee59a195a9ca14b","subject":"Merge \"New release model: correct the named branches lifetime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/4c616040db7fd135f348f09bcee59a195a9ca14b"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-10-27 10:54:08.000000000","tz":0},"committer":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-10-27 14:07:00.000000000","tz":0},"subject":"Add PXE boot from iSCSI volume scenario","message":"Add PXE boot from iSCSI volume scenario\n\nStory: #2008023\nTask: #41164\nChange-Id: I3849e4e644a6e049ca48a734e45e70f35181afbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/82c4fc46f7c8d4f3c66f668931d1fd3af8887a06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/82c4fc46f7c8d4f3c66f668931d1fd3af8887a06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
