)]}'
{"id":"openstack%2Fironic-specs~960550","triplet_id":"openstack%2Fironic-specs~master~I9a5854ce81b0a1478af34604b4fe093ecdac6e01","project":"openstack/ironic-specs","branch":"master","topic":"feature/standalone-networking","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-11-04 10:25:50.000000000","reason":"Change was submitted"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-10-03 16:20:02.000000000","reason":"\u003cGERRIT_ACCOUNT_11655\u003e replied on the change","reason_account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},"25437":{"account":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"last_update":"2025-11-04 10:25:50.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-11-04 10:25:50.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I9a5854ce81b0a1478af34604b4fe093ecdac6e01","subject":"Update standalone networking initial conditions","status":"MERGED","created":"2025-09-11 13:06:18.000000000","updated":"2025-11-04 10:27:04.000000000","submitted":"2025-11-04 10:25:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"960550-feature/standalone-networking","meta_rev_id":"8152a98c44e6182e713bc5a8706990bf747c366b","_number":960550,"virtual_id_number":960550,"owner":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-04 10:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2025-10-03 16:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2025-11-04 09:58:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2025-11-04 09:58:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-11 13:07:25.000000000","updated_by":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2025-09-11 13:15:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-15 17:59:30.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2025-09-23 16:28:44.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2025-09-24 12:24:37.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2025-11-04 09:58:41.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"68f566770df52aad2c79485c8eba3d65e2bbae4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-09-11 13:06:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a103e0ac28f2663480617bbd2fa418d788f4f6aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 13:15:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63a29804483d4352ba6e8dd58eb399bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cadf6094370546b6889d884d9933d4e4 : SUCCESS in 7m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/62e5951f056647adac841d7dacbba38b : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b86641540da8e3cdc7848addc81f6da288455dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-09-11 14:53:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"412abc01789feb34989bd3ca2d39d372a16b058e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 15:04:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c18082878a7e4ef6a3ac58b5bd0d610c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf9d4130381a457b81955b36bd26c289 : SUCCESS in 8m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e0271cf98ea345b1b085e393b62a2a58 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"14e60ba2cbb707ac5c25e4df4591713bbaa3f07f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-09-15 15:38:08.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"bafb9f010f1a1d505423cc74151bc63c0f98a561","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-09-15 17:59:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eec2b20b8aa604670e64007c7c426252d4714d15","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-09-15 18:23:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a0525137b3b9e54021e8127996e16b83131702d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-09-22 12:24:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"87db23233cdfdb0ef2713c9938f765c3d49d193e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 12:40:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1451dcac5a7544d89c9e2110512300c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea6583fd1981435693654c9c41823506 : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/498b430cfa4c41b980ca784e0a78bff1 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9661f7b248fce35b1a4b08c6a87bfd2080ada47a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-09-23 16:28:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b8c60c2b5e2359c6ad485cbaa5bba36d3f6ff98d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-09-23 16:45:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"87e6beb20de9196bf5b7ac4d75ccf464883cceef","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-09-23 17:46:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"83997d1348b0cdcc5efbd29cd3ef02a8c329b525","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-09-24 12:24:37.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"603ef5d70d22ae3cb6eed12540d5ba28ec23273c","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-09-24 18:11:19.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"737a03ae244ef8c813df7b0d1c303c18617cc5d5","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-01 13:06:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"98ee1ed2b4511933f8322074f029da8af83f3249","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-01 17:48:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d921d5541d48006741a63234a7f1ece2aece602e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 17:57:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/445663cb505a4f3f9f38fdce2d2ca07e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f552f3a8411e4cfca6ad7b81e9c5193e : SUCCESS in 7m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f667cecb10e144b88ad380d0193d0e60 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":4},{"id":"47a903d2bfa47e265da8e227606c4acc84ab2b64","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-02 00:24:04.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cecb0d0d69a0b17d6be5d50b6324f1fbfd8330db","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-10-02 13:06:54.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3bc7585a035148314b3535fc868be46254cf74fd","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-02 13:11:11.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"44f12779d5c0a442d3b134edc8a2e003a18dbb23","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-10-02 13:46:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"93a26d7491ddac7e4342a840766bb031121dd30b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-02 20:37:48.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f1560abc6e923e890a6a834590230fdecac981db","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-02 20:38:59.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"28dd9a7852525ef913ff3936b005d8a8f7e91238","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-02 20:39:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b002cb4effb9757233490324d857003d99f171a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 20:45:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59c1399c55be4b6798d26789c6e0d729\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad91ecc71ff14bd2bd81956f1286b764 : SUCCESS in 6m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cf44de8ecab64837a5f3706d2a69a1f4 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":5},{"id":"047e4251d16bcf46e69b2a3b0cfea699bce24fab","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-10-03 16:20:02.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c190c67b4ebef46e75d4085b540191b2e71ec73a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-04 09:58:41.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a3fcf53601725463498850eccbec745b0e5a7af7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 09:58:50.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2321b23938d8fb5d63a2df59b1cda70b33069656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 10:25:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bffcb9ac5977491083b0eb30785ed3e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9e5fa3a2274d92bfd6d1c0322e6922 : SUCCESS in 9m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/452d998b5588485288c9b55b04a4b8c4 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":5},{"id":"4bb4c816dace4b623e7eaa118774dce78cbfc93d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 10:25:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"8152a98c44e6182e713bc5a8706990bf747c366b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 10:27:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05e0b7937ad14f1f98f9f714153fa010\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e9dc7439eecb4b82956928a311665ae1 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f3fd4d7e87bad6cbef1b9960b453c45bd94c2b9d","revisions":{"c1ea8839193f25d776f1b02213542d5f5e58fe6a":{"kind":"REWORK","_number":1,"created":"2025-09-11 13:06:18.000000000","uploader":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"ref":"refs/changes/50/960550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/50/960550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/1"}}},"commit":{"parents":[{"commit":"0c084d53f7db4fa5d7ffc73da2e735326a660b15","subject":"Merge \"Add support for standalone network configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/0c084d53f7db4fa5d7ffc73da2e735326a660b15"}]}],"author":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-09-11 12:59:05.000000000","tz":-240},"committer":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-09-11 13:06:07.000000000","tz":-240},"subject":"Update standalone networking initial conditions","message":"Update standalone networking initial conditions\n\nThis updates the standalone networking spec to addresses issues related\nto the initial starting conditions.  Specifically, there is a \u0027chicken-\nand-egg\u0027 scenario that exists in that LLDP information is required\nin order to know which switch ports need to be configured for a node\nbut if those switch ports are disabled by default there is no way to\nknow which switch ports to configure.\n\nThis updates the document to state that switch ports must be manually\nconfigured ahead of time and must be setup on the inspection VLAN to\nfacilitate collecting LLDP information.\n\nRelated-Bug: 2113769\nChange-Id: I9a5854ce81b0a1478af34604b4fe093ecdac6e01\nSigned-off-by: Allain Legacy \u003calegacy@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/c1ea8839193f25d776f1b02213542d5f5e58fe6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/c1ea8839193f25d776f1b02213542d5f5e58fe6a"}]},"branch":"refs/heads/master"},"2589cd369b5be0a55ebb7a2debf55a9d44dfab61":{"kind":"REWORK","_number":2,"created":"2025-09-11 14:53:25.000000000","uploader":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"ref":"refs/changes/50/960550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/50/960550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/2"}}},"commit":{"parents":[{"commit":"0c084d53f7db4fa5d7ffc73da2e735326a660b15","subject":"Merge \"Add support for standalone network configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/0c084d53f7db4fa5d7ffc73da2e735326a660b15"}]}],"author":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-09-11 12:59:05.000000000","tz":-240},"committer":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-09-11 14:53:19.000000000","tz":-240},"subject":"Update standalone networking initial conditions","message":"Update standalone networking initial conditions\n\nThis updates the standalone networking spec to addresses issues related\nto the initial starting conditions.  Specifically, there is a \u0027chicken-\nand-egg\u0027 scenario that exists in that LLDP information is required\nin order to know which switch ports need to be configured for a node\nbut if those switch ports are disabled by default there is no way to\nknow which switch ports to configure.\n\nThis updates the document to state that switch ports must be manually\nconfigured ahead of time and must be setup on the inspection VLAN to\nfacilitate collecting LLDP information.\n\nRelated-Bug: 2113769\nChange-Id: I9a5854ce81b0a1478af34604b4fe093ecdac6e01\nSigned-off-by: Allain Legacy \u003calegacy@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/2589cd369b5be0a55ebb7a2debf55a9d44dfab61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/2589cd369b5be0a55ebb7a2debf55a9d44dfab61"}]},"branch":"refs/heads/master"},"c914511db399ee1c0efa2d02d322722558375852":{"kind":"REWORK","_number":3,"created":"2025-09-22 12:24:51.000000000","uploader":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"ref":"refs/changes/50/960550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/50/960550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/3"}}},"commit":{"parents":[{"commit":"0c084d53f7db4fa5d7ffc73da2e735326a660b15","subject":"Merge \"Add support for standalone network configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/0c084d53f7db4fa5d7ffc73da2e735326a660b15"}]}],"author":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-09-11 12:59:05.000000000","tz":-240},"committer":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-09-22 12:24:39.000000000","tz":-240},"subject":"Update standalone networking initial conditions","message":"Update standalone networking initial conditions\n\nThis updates the standalone networking spec to addresses issues related\nto the initial starting conditions.  Specifically, there is a \u0027chicken-\nand-egg\u0027 scenario that exists in that LLDP information is required\nin order to know which switch ports need to be configured for a node\nbut if those switch ports are disabled by default there is no way to\nknow which switch ports to configure.\n\nThis updates the document to state that switch ports must be manually\nconfigured ahead of time and must be setup on the inspection VLAN to\nfacilitate collecting LLDP information.\n\nRelated-Bug: 2113769\nChange-Id: I9a5854ce81b0a1478af34604b4fe093ecdac6e01\nSigned-off-by: Allain Legacy \u003calegacy@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/c914511db399ee1c0efa2d02d322722558375852"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/c914511db399ee1c0efa2d02d322722558375852"}]},"branch":"refs/heads/master"},"8527d63663abe63239df31ac8d9e475f1724d5c1":{"kind":"REWORK","_number":4,"created":"2025-10-01 17:48:57.000000000","uploader":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"ref":"refs/changes/50/960550/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/50/960550/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/4"}}},"commit":{"parents":[{"commit":"0c084d53f7db4fa5d7ffc73da2e735326a660b15","subject":"Merge \"Add support for standalone network configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/0c084d53f7db4fa5d7ffc73da2e735326a660b15"}]}],"author":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-09-11 12:59:05.000000000","tz":-240},"committer":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-10-01 17:48:41.000000000","tz":-240},"subject":"Update standalone networking initial conditions","message":"Update standalone networking initial conditions\n\nThis updates the standalone networking spec to addresses issues related\nto the initial starting conditions.  Specifically, there is a \u0027chicken-\nand-egg\u0027 scenario that exists in that LLDP information is required\nin order to know which switch ports need to be configured for a node\nbut if those switch ports are disabled by default there is no way to\nknow which switch ports to configure.\n\nThis updates the document to state that switch ports must be manually\nconfigured ahead of time and must be setup on the inspection VLAN to\nfacilitate collecting LLDP information.\n\nRelated-Bug: 2113769\nChange-Id: I9a5854ce81b0a1478af34604b4fe093ecdac6e01\nSigned-off-by: Allain Legacy \u003calegacy@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/8527d63663abe63239df31ac8d9e475f1724d5c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/8527d63663abe63239df31ac8d9e475f1724d5c1"}]},"branch":"refs/heads/master"},"f3fd4d7e87bad6cbef1b9960b453c45bd94c2b9d":{"kind":"REWORK","_number":5,"created":"2025-10-02 20:37:48.000000000","uploader":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"ref":"refs/changes/50/960550/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-specs","ref":"refs/changes/50/960550/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-specs refs/changes/50/960550/5"}}},"commit":{"parents":[{"commit":"0c084d53f7db4fa5d7ffc73da2e735326a660b15","subject":"Merge \"Add support for standalone network configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/0c084d53f7db4fa5d7ffc73da2e735326a660b15"}]}],"author":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-09-11 12:59:05.000000000","tz":-240},"committer":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-10-02 20:37:38.000000000","tz":-240},"subject":"Update standalone networking initial conditions","message":"Update standalone networking initial conditions\n\nThis updates the standalone networking spec to addresses issues related\nto the initial starting conditions.  Specifically, there is a \u0027chicken-\nand-egg\u0027 scenario that exists in that LLDP information is required\nin order to know which switch ports need to be configured for a node\nbut if those switch ports are disabled by default there is no way to\nknow which switch ports to configure.\n\nThis updates the document to state that switch ports must be manually\nconfigured ahead of time and must be setup on the inspection VLAN to\nfacilitate collecting LLDP information.\n\nRelated-Bug: 2113769\nChange-Id: I9a5854ce81b0a1478af34604b4fe093ecdac6e01\nSigned-off-by: Allain Legacy \u003calegacy@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/f3fd4d7e87bad6cbef1b9960b453c45bd94c2b9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-specs/commit/f3fd4d7e87bad6cbef1b9960b453c45bd94c2b9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
