)]}'
{"id":"openstack%2Fironic-tempest-plugin~489784","triplet_id":"openstack%2Fironic-tempest-plugin~master~Icd61bbbdc0f0633a93adef89652d54a87d835534","project":"openstack/ironic-tempest-plugin","branch":"master","hashtags":[],"change_id":"Icd61bbbdc0f0633a93adef89652d54a87d835534","subject":"Fix BaseBaremetalTest._assertExpected docstring","status":"ABANDONED","created":"2017-08-02 00:07:56.000000000","updated":"2017-08-02 16:26:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"483c7c94ce9c4383374bbf0b9bdbc5fc98649920","_number":489784,"virtual_id_number":489784,"owner":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":-1,"date":"2017-08-02 00:40:15.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2017-08-02 00:07:59.000000000","_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-02 00:07:59.000000000","updated_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"reviewer":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"state":"REVIEWER"},{"updated":"2017-08-02 00:40:15.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"76eab045d141ee5f81702748e956b4688d1c40a9","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2017-08-02 00:07:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49fbdd8aae84fcdff2eab568b5b3579104aaf169","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-02 00:40:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-ironic-inspector-tempest-plugin-pep8-ubuntu-xenial http://logs.openstack.org/84/489784/1/check/gate-ironic-inspector-tempest-plugin-pep8-ubuntu-xenial/83c0593/ : FAILURE in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"483c7c94ce9c4383374bbf0b9bdbc5fc98649920","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2017-08-02 16:26:17.000000000","message":"Abandoned\n\nWill work with infra to do a batch upload.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"304a73bbaba4f45f03a24ec158776865810b89a7","revisions":{"304a73bbaba4f45f03a24ec158776865810b89a7":{"kind":"REWORK","_number":1,"created":"2017-08-02 00:07:56.000000000","uploader":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"ref":"refs/changes/84/489784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/84/489784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/84/489784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/84/489784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/84/489784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/84/489784/1"}}},"commit":{"parents":[{"commit":"edbad37bbfabc397b59cd9ab0f71c076512c302a","subject":"Deduplicate _assertExpected method in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/edbad37bbfabc397b59cd9ab0f71c076512c302a"}]}],"author":{"name":"Ruby Loo","email":"ruby.loo@intel.com","date":"2017-02-27 15:45:03.000000000","tz":0},"committer":{"name":"John L. Villalovos","email":"john.l.villalovos@intel.com","date":"2017-08-01 23:45:50.000000000","tz":-420},"subject":"Fix BaseBaremetalTest._assertExpected docstring","message":"Fix BaseBaremetalTest._assertExpected docstring\n\nThis updates the docstring for tempest plugin\u0027s\nBaseBaremetalTest._assertExpected() so that it clarifies\nwhat it is doing and what the parameters are.\n\nChange-Id: Icd61bbbdc0f0633a93adef89652d54a87d835534\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/304a73bbaba4f45f03a24ec158776865810b89a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/304a73bbaba4f45f03a24ec158776865810b89a7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"edbad37bbfabc397b59cd9ab0f71c076512c302a","is_merged_in_target_branch":false,"change_id":"Icba38b77529c78895d713a7ee7bf19193a8141cf","change_number":489783,"patch_set_number":1,"change_status":"ABANDONED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
