)]}'
{"id":"openstack%2Fironic-tempest-plugin~489793","triplet_id":"openstack%2Fironic-tempest-plugin~master~I3ca7cf3af42e0e55431d5affa36e567f147aa850","project":"openstack/ironic-tempest-plugin","branch":"master","hashtags":[],"change_id":"I3ca7cf3af42e0e55431d5affa36e567f147aa850","subject":"Prune local copy of tempest.scenario.manager.py","status":"ABANDONED","created":"2017-08-02 00:07:56.000000000","updated":"2017-08-02 16:26:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f17e6cbea182538abbec4de2fc6a60951837725f","_number":489793,"virtual_id_number":489793,"owner":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":-1,"date":"2017-08-02 00:44:49.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":22860,"name":"Solio Sarabia","email":"solio.sarabia@intel.com","username":"ssarabia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22860,"name":"Solio Sarabia","email":"solio.sarabia@intel.com","username":"ssarabia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2017-08-02 00:07:59.000000000","_account_id":22860,"name":"Solio Sarabia","email":"solio.sarabia@intel.com","username":"ssarabia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":22860,"name":"Solio Sarabia","email":"solio.sarabia@intel.com","username":"ssarabia"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":22860,"name":"Solio Sarabia","email":"solio.sarabia@intel.com","username":"ssarabia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-02 00:07:59.000000000","updated_by":{"_account_id":22860,"name":"Solio Sarabia","email":"solio.sarabia@intel.com","username":"ssarabia"},"reviewer":{"_account_id":22860,"name":"Solio Sarabia","email":"solio.sarabia@intel.com","username":"ssarabia"},"state":"REVIEWER"},{"updated":"2017-08-02 00:44:49.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"e996fae67f99844a2c51b76af1486d231888d375","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2017-08-02 00:07:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d3b5f6014de829b867fa36b70105a6b99befd98","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-02 00:44:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-ironic-inspector-tempest-plugin-pep8-ubuntu-xenial http://logs.openstack.org/93/489793/1/check/gate-ironic-inspector-tempest-plugin-pep8-ubuntu-xenial/2087252/ : FAILURE in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f17e6cbea182538abbec4de2fc6a60951837725f","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2017-08-02 16:26:14.000000000","message":"Abandoned\n\nWill work with infra to do a batch upload.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab9ddd688a7f388573dca29e83a9353550ee394a","revisions":{"ab9ddd688a7f388573dca29e83a9353550ee394a":{"kind":"REWORK","_number":1,"created":"2017-08-02 00:07:56.000000000","uploader":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"ref":"refs/changes/93/489793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/93/489793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/93/489793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/93/489793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/93/489793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/93/489793/1"}}},"commit":{"parents":[{"commit":"c65742a8c90e26895e221f1e61d22431383d094a","subject":"Include a copy of tempest.scenario.manager module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/c65742a8c90e26895e221f1e61d22431383d094a"}]}],"author":{"name":"Solio Sarabia","email":"solio.sarabia@intel.com","date":"2017-03-17 04:28:52.000000000","tz":-300},"committer":{"name":"John L. Villalovos","email":"john.l.villalovos@intel.com","date":"2017-08-01 23:59:46.000000000","tz":-420},"subject":"Prune local copy of tempest.scenario.manager.py","message":"Prune local copy of tempest.scenario.manager.py\n\nRemove unused code from manager.py to better identify ironic\ndependencies on this tempest interface. This is a follow-up\npatch to https://review.openstack.org/#/c/439252/.\n\nChange-Id: I3ca7cf3af42e0e55431d5affa36e567f147aa850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/ab9ddd688a7f388573dca29e83a9353550ee394a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/ab9ddd688a7f388573dca29e83a9353550ee394a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c65742a8c90e26895e221f1e61d22431383d094a","is_merged_in_target_branch":false,"change_id":"Iaa06df6ab75fb8f79cbcd3f6b21e68622e7799c1","change_number":489792,"patch_set_number":1,"change_status":"ABANDONED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
