)]}'
{"id":"openstack%2Fironic-tempest-plugin~637552","triplet_id":"openstack%2Fironic-tempest-plugin~master~If1c61a4d40f83a8b15462ec7890267b9254b03d7","project":"openstack/ironic-tempest-plugin","branch":"master","topic":"test_rocky","hashtags":[],"change_id":"If1c61a4d40f83a8b15462ec7890267b9254b03d7","subject":"DNM - Test","status":"ABANDONED","created":"2019-02-18 14:16:42.000000000","updated":"2019-02-21 16:32:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bf783566f29974f418fe48f54cb4a8f8399f35d6","_number":637552,"virtual_id_number":637552,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-02-18 16:32:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-18 16:32:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"664a8daf2849e0c70dcc694d636f648789d4aa58","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-02-18 14:16:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b397500388efd070784051ff052cb8f4a99d7ed3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-18 15:17:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/52/637552/1/check/openstack-tox-docs/b7c4c8e/html/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/52/637552/1/check/openstack-tox-pep8/0c32721/ : SUCCESS in 4m 11s\n- ironic-standalone http://logs.openstack.org/52/637552/1/check/ironic-standalone/2b4a374/ : SUCCESS in 51m 28s\n- ironic-tempest-functional-python3 http://logs.openstack.org/52/637552/1/check/ironic-tempest-functional-python3/39e06d9/ : SUCCESS in 24m 13s\n- ironic-inspector-tempest http://logs.openstack.org/52/637552/1/check/ironic-inspector-tempest/c9d9741/ : SUCCESS in 51m 56s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode http://logs.openstack.org/52/637552/1/check/ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode/b496b55/ : SUCCESS in 57m 54s\n- ironic-inspector-tempest-dsvm-discovery-rocky http://logs.openstack.org/52/637552/1/check/ironic-inspector-tempest-dsvm-discovery-rocky/43f47ca/ : FAILURE in 57m 29s (non-voting)\n- ironic-inspector-tempest-dsvm-discovery-queens http://logs.openstack.org/52/637552/1/check/ironic-inspector-tempest-dsvm-discovery-queens/f5b94bc/ : SUCCESS in 41m 28s (non-voting)\n- ironic-inspector-tempest-dsvm-discovery-pike http://logs.openstack.org/52/637552/1/check/ironic-inspector-tempest-dsvm-discovery-pike/57f0b41/ : SUCCESS in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78800b29c47116674b2a0faded8b164cc7a97e8b","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-02-18 15:30:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bef1a66196ffa76f6f0fc9e538a351aeac42ceb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-18 16:32:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/52/637552/2/check/openstack-tox-docs/0ce9e38/html/ : SUCCESS in 5m 55s\n- openstack-tox-pep8 http://logs.openstack.org/52/637552/2/check/openstack-tox-pep8/bacf849/ : SUCCESS in 4m 13s\n- ironic-standalone http://logs.openstack.org/52/637552/2/check/ironic-standalone/6bd6cb0/ : SUCCESS in 53m 18s\n- ironic-tempest-functional-python3 http://logs.openstack.org/52/637552/2/check/ironic-tempest-functional-python3/5a25b3c/ : SUCCESS in 25m 06s\n- ironic-inspector-tempest http://logs.openstack.org/52/637552/2/check/ironic-inspector-tempest/6a784f3/ : SUCCESS in 46m 25s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode http://logs.openstack.org/52/637552/2/check/ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode/a5bbc4c/ : SUCCESS in 57m 56s\n- ironic-inspector-tempest-dsvm-discovery-rocky http://logs.openstack.org/52/637552/2/check/ironic-inspector-tempest-dsvm-discovery-rocky/7e8e604/ : SUCCESS in 45m 56s (non-voting)\n- ironic-inspector-tempest-dsvm-discovery-queens http://logs.openstack.org/52/637552/2/check/ironic-inspector-tempest-dsvm-discovery-queens/ad5db4b/ : SUCCESS in 45m 06s (non-voting)\n- ironic-inspector-tempest-dsvm-discovery-pike http://logs.openstack.org/52/637552/2/check/ironic-inspector-tempest-dsvm-discovery-pike/01a8f66/ : SUCCESS in 45m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"71d8a355d66d316cad3672af63524467e6e4e499","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-02-21 16:32:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"581f923831061795b01711b4d4bd62f3861ce31e","revisions":{"97bb27468fe92daf9083fc783ff4514cbec0aa5a":{"kind":"REWORK","_number":1,"created":"2019-02-18 14:16:42.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/52/637552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/52/637552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/52/637552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/52/637552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/52/637552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/52/637552/1"}}},"commit":{"parents":[{"commit":"06ff9cb3d13d6ffdd927bc196f10e2ca8019cd2c","subject":"Merge \"Allocation API: functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/06ff9cb3d13d6ffdd927bc196f10e2ca8019cd2c"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-02-18 14:16:11.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-02-18 14:16:22.000000000","tz":60},"subject":"DNM - Test","message":"DNM - Test\n\nChange-Id: If1c61a4d40f83a8b15462ec7890267b9254b03d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/97bb27468fe92daf9083fc783ff4514cbec0aa5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/97bb27468fe92daf9083fc783ff4514cbec0aa5a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06ff9cb3d13d6ffdd927bc196f10e2ca8019cd2c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"581f923831061795b01711b4d4bd62f3861ce31e":{"kind":"REWORK","_number":2,"created":"2019-02-18 15:30:55.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/52/637552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/52/637552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/52/637552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/52/637552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/52/637552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/52/637552/2"}}},"commit":{"parents":[{"commit":"06ff9cb3d13d6ffdd927bc196f10e2ca8019cd2c","subject":"Merge \"Allocation API: functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/06ff9cb3d13d6ffdd927bc196f10e2ca8019cd2c"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-02-18 14:16:11.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-02-18 15:30:14.000000000","tz":60},"subject":"DNM - Test","message":"DNM - Test\n\nDepends-On: https://review.openstack.org/#/c/635822/\nChange-Id: If1c61a4d40f83a8b15462ec7890267b9254b03d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/581f923831061795b01711b4d4bd62f3861ce31e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/581f923831061795b01711b4d4bd62f3861ce31e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06ff9cb3d13d6ffdd927bc196f10e2ca8019cd2c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
