)]}'
{"id":"openstack%2Fironic-tempest-plugin~687551","triplet_id":"openstack%2Fironic-tempest-plugin~master~I654efd99595570a8bb5a0fdddb17c24408c32f55","project":"openstack/ironic-tempest-plugin","branch":"master","topic":"fix_py3_standalone","hashtags":[],"change_id":"I654efd99595570a8bb5a0fdddb17c24408c32f55","subject":"Fix tests for ironic-standalone with py3","status":"ABANDONED","created":"2019-10-09 13:15:51.000000000","updated":"2019-10-23 15:14:47.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"929acc3e22e3f1c707583d665cf1e3f791b87bdb","_number":687551,"virtual_id_number":687551,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-09 13:20:19.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-09 15:11:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-09 14:31:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 13:20:19.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-10-09 15:11:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6da0426b58d09be1a7ff03a87eb1d3ad54e7e6a6","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-10-09 13:15:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"497343b8ee6e32cdd9aede924b60e915938e99e8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-09 13:20:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1314da45dc1211b5a9a7e8394a11ee310449c9e8","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-10-09 14:31:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a8a476d37480822b2e93a72fce83fd9ea7199fa","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-10-09 14:31:56.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7c936ac9d46e0701c349565c429829aa3c1a46b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:11:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/979d98c0d4444380a3ca35ac154cfc09 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b94e2ecb694a6da87a43578ec2c1e3 : SUCCESS in 5m 05s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/6c39897f5b4647d09decc20e63941ac4 : SUCCESS in 1h 24m 06s\n- ironic-standalone-train https://zuul.opendev.org/t/openstack/build/539117a9061740bf97a6fd8445a3dbb7 : SUCCESS in 1h 10m 16s\n- ironic-standalone-stein https://zuul.opendev.org/t/openstack/build/9960cfebbc384b139788d9f3fa877d1f : SUCCESS in 1h 04m 52s\n- ironic-dsvm-standalone-rocky https://zuul.opendev.org/t/openstack/build/3f68a7f177fc43d5897e3179403312b8 : SUCCESS in 1h 00m 17s (non-voting)\n- ironic-dsvm-standalone-queens https://zuul.opendev.org/t/openstack/build/16b083fcece34c1b85e2315c947d25b2 : FAILURE in 1h 13m 24s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/78eafaa24cb9430a86e233d03d3c93bf : SUCCESS in 24m 04s\n- ironic-tempest-functional-python3-train https://zuul.opendev.org/t/openstack/build/0d812edd942744528b8ff74b3a4d1b93 : SUCCESS in 29m 03s\n- ironic-tempest-functional-python3-stein https://zuul.opendev.org/t/openstack/build/5d95f5c58abf4bfca3f539a80f7de729 : SUCCESS in 27m 45s\n- ironic-tempest-dsvm-functional-python3-rocky https://zuul.opendev.org/t/openstack/build/a2d346c306644f239c58b02e2a6782c3 : SUCCESS in 23m 41s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/56d1560eab7044cbbf2e9b8eac6eec5e : SUCCESS in 40m 49s\n- ironic-inspector-tempest-train https://zuul.opendev.org/t/openstack/build/678cd7440c604bcc82a83089ebf8f581 : SUCCESS in 55m 00s\n- ironic-inspector-tempest-stein https://zuul.opendev.org/t/openstack/build/f489a63aa3624ea88d4e82e4ed397c8b : SUCCESS in 52m 28s\n- ironic-tempest-dsvm-ironic-inspector-rocky https://zuul.opendev.org/t/openstack/build/3a957b504b204b4898bc2bfdd872af3d : SUCCESS in 55m 57s (non-voting)\n- ironic-tempest-dsvm-ironic-inspector-queens https://zuul.opendev.org/t/openstack/build/e6d6623d5a2545ca80ae6f8aed885f03 : FAILURE in 1h 18m 52s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/7dbe2c8221bc45d19fca734dbe885a51 : SUCCESS in 1h 36m 30s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-train https://zuul.opendev.org/t/openstack/build/b9052d7872be4dfeb1715b4300591e2d : SUCCESS in 1h 40m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-stein https://zuul.opendev.org/t/openstack/build/92dcf9e4bdfb4e8795a1fe7b3a75278b : SUCCESS in 1h 49m 05s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode-rocky https://zuul.opendev.org/t/openstack/build/0d2a4521a7c1496fa56a63aa3e5e6376 : FAILURE in 1h 27m 29s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode-queens https://zuul.opendev.org/t/openstack/build/1a0195297ecb4803aab47e9d7ab5f683 : FAILURE in 47m 13s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/6869bee36a4943a386971b6ad4747bec : SUCCESS in 48m 25s\n- ironic-inspector-tempest-discovery-train https://zuul.opendev.org/t/openstack/build/f9b72c22b1c94cd5ab776e1c38683839 : SUCCESS in 46m 24s (non-voting)\n- ironic-inspector-tempest-discovery-stein https://zuul.opendev.org/t/openstack/build/e4f4353d545c4e64bc6dbd17e8d1dc05 : FAILURE in 43m 26s (non-voting)\n- ironic-inspector-tempest-dsvm-discovery-rocky https://zuul.opendev.org/t/openstack/build/ed56c323adcf4923859f3d3a5b0b5e62 : SUCCESS in 43m 23s (non-voting)\n- ironic-inspector-tempest-dsvm-discovery-queens https://zuul.opendev.org/t/openstack/build/e7057678a95545e5a842b24820c0e1b1 : SUCCESS in 40m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b238a74436a01375eff369147a4ee466a869e1e1","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-10-23 15:14:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5519d8f02dc11ae4509097d2f2cc1a6247ce624","revisions":{"e5519d8f02dc11ae4509097d2f2cc1a6247ce624":{"kind":"REWORK","_number":1,"created":"2019-10-09 13:15:51.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/51/687551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/51/687551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/51/687551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/51/687551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/51/687551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/51/687551/1"}}},"commit":{"parents":[{"commit":"ed09bccbae801f1d9b54054b72157aea24f1af5b","subject":"Merge \"Add CI jobs for Train stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/ed09bccbae801f1d9b54054b72157aea24f1af5b"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-10-09 12:43:43.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2019-10-09 12:46:40.000000000","tz":120},"subject":"Fix tests for ironic-standalone with py3","message":"Fix tests for ironic-standalone with py3\n\nWhen running the ironic-standalone job using python3\ntempest fails with `UnboundLocalError` [1]\n\n[1] https://zuul.opendev.org/t/openstack/build/1e83c2c43e6c4391822e5ce37d6053e8\n\nChange-Id: I654efd99595570a8bb5a0fdddb17c24408c32f55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/e5519d8f02dc11ae4509097d2f2cc1a6247ce624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/e5519d8f02dc11ae4509097d2f2cc1a6247ce624"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
