)]}'
{"id":"openstack%2Fironic-tempest-plugin~711054","triplet_id":"openstack%2Fironic-tempest-plugin~master~I5b619321be876e04f4b29dce083ca0cc80c4ed5b","project":"openstack/ironic-tempest-plugin","branch":"master","topic":"use_get_node_vifs_only","hashtags":[],"change_id":"I5b619321be876e04f4b29dce083ca0cc80c4ed5b","subject":"Refactor validate_ports function","status":"MERGED","created":"2020-03-03 16:46:23.000000000","updated":"2020-03-04 10:02:48.000000000","submitted":"2020-03-04 09:59:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711054-1583315990507-20d59637","meta_rev_id":"1ecfe77d9f6a5db006fc5197d168dcda3922a02e","_number":711054,"virtual_id_number":711054,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 09:59:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-03-04 08:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-03 16:57:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2020-03-04 08:20:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 16:57:21.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-03-04 08:20:55.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-03-04 09:59:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b3b08a55e633bca4b7a0f7cd1678df522c632cb","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-03-03 16:46:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ac9e123f3b609966f3f2b664bc0634b2297c7f0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-03 16:57:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc4d1ddd0162222a011f6e3ab0f5983a34839518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 22:12:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e48152cf37423c9c9b36d99f6ca0b7 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebf5203ceb234a789b6dffddae582642 : SUCCESS in 4m 36s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/29ab8d9c03f649f592c5d1842245690a : SUCCESS in 1h 31m 39s\n- ironic-standalone-train https://zuul.opendev.org/t/openstack/build/d8207eeba1774bb9843e88da9d7af98c : SUCCESS in 1h 44m 49s\n- ironic-standalone-stein https://zuul.opendev.org/t/openstack/build/08f9c0850d5745aa8bc721cf94970ed8 : SUCCESS in 56m 06s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/d34fbf343e3841d2af0ebe39b3ff70ce : SUCCESS in 33m 09s\n- ironic-tempest-functional-python3-train https://zuul.opendev.org/t/openstack/build/381775bcbacb4a5aa84f839764b11eae : SUCCESS in 30m 50s\n- ironic-tempest-functional-python3-stein https://zuul.opendev.org/t/openstack/build/0d922a3ec3f84881a83f53051fbe0e79 : SUCCESS in 29m 48s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/7e8976458787491c9d46fcd84cfb0aea : SUCCESS in 55m 05s\n- ironic-inspector-tempest-train https://zuul.opendev.org/t/openstack/build/08f0c534381e4cbba4c9f289cc12e68a : SUCCESS in 53m 31s\n- ironic-inspector-tempest-stein https://zuul.opendev.org/t/openstack/build/101bbce5a292460d93a85b2c515fe693 : SUCCESS in 48m 29s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/3c8c658c425648eb9d03f4cc038839d4 : FAILURE in 3h 02m 32s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-train https://zuul.opendev.org/t/openstack/build/37c13a38b0ff4c43acd24917d9d2cf08 : SUCCESS in 1h 52m 49s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-stein https://zuul.opendev.org/t/openstack/build/0d0d27708e5f428098bed2d8547e1edf : FAILURE in 1h 36m 08s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/15cabcbb537d45849a6817c16d468bed : SUCCESS in 45m 29s\n- ironic-inspector-tempest-discovery-train https://zuul.opendev.org/t/openstack/build/74d5234fbf3f41019c127fe908c8c014 : SUCCESS in 40m 51s (non-voting)\n- ironic-inspector-tempest-discovery-stein https://zuul.opendev.org/t/openstack/build/11144082ee334992b6db1300ac4499a7 : SUCCESS in 47m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d3c38df5b5d3198af0119b48249162ec462f379","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-04 08:20:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eeb2efad492aacafe0f336f3e7f5fb1f86f6b6b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 08:21:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8bde71a8e687dd84435c88ab6ec41643ce1b0f48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 09:59:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9237bdb0fedfb12e82613a2fae484edd4bfaa55e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 09:59:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f273c0394e45a79c5552b51420445a : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a7c6fde404b486f991a3a75969cecfe : SUCCESS in 3m 45s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/be339b72713e43698505f81e77715c2b : SUCCESS in 1h 37m 20s\n- ironic-standalone-train https://zuul.opendev.org/t/openstack/build/8fb94aca64274815b4e081cb5a0091c9 : SUCCESS in 1h 15m 02s\n- ironic-standalone-stein https://zuul.opendev.org/t/openstack/build/d6d6670888474dc79749228325178678 : SUCCESS in 55m 49s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/36ab5da33465457ca4f115ee8d865fcc : SUCCESS in 29m 42s\n- ironic-tempest-functional-python3-train https://zuul.opendev.org/t/openstack/build/92fd332a15d54ba284508f18a723db47 : SUCCESS in 28m 24s\n- ironic-tempest-functional-python3-stein https://zuul.opendev.org/t/openstack/build/5ab2a1a9d6cf4fb49b9600c34418c2e6 : SUCCESS in 27m 57s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/f15ebdb2a7294513a829c766cf7d40a1 : SUCCESS in 55m 53s\n- ironic-inspector-tempest-train https://zuul.opendev.org/t/openstack/build/4de486cafcdf44a0b85f818e3f9dbe3a : SUCCESS in 55m 31s\n- ironic-inspector-tempest-stein https://zuul.opendev.org/t/openstack/build/fb732d5c972149ab8c1a8cf786036e3d : SUCCESS in 1h 00m 01s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/a4ae8333a50449ad95d96e159437b8d5 : SUCCESS in 47m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6a96e6cbd8f48ddd60053bc1baf458b1b7c39e47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 10:02:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/914712f65e094bb1bf8c17968b331c9e : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"728940fd30f6121e5b7bbfb895d3c43f9412151d","revisions":{"728940fd30f6121e5b7bbfb895d3c43f9412151d":{"kind":"REWORK","_number":1,"created":"2020-03-03 16:46:23.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/54/711054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/54/711054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/711054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/711054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/711054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/711054/1"}}},"commit":{"parents":[{"commit":"790bf18d2e5c25dc360cfbd3aa3bc7869adadf5d","subject":"Remove the SW RAID configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/790bf18d2e5c25dc360cfbd3aa3bc7869adadf5d"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-03-03 16:43:26.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-03-03 16:43:26.000000000","tz":60},"subject":"Refactor validate_ports function","message":"Refactor validate_ports function\n\nSince 1.28 is ancient and all the stable branches have support\nfor `get_node_vifs` let\u0027s remove the unnecessary check.\n\nChange-Id: I5b619321be876e04f4b29dce083ca0cc80c4ed5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/728940fd30f6121e5b7bbfb895d3c43f9412151d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/728940fd30f6121e5b7bbfb895d3c43f9412151d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
