)]}'
{"id":"openstack%2Fironic-tempest-plugin~913926","triplet_id":"openstack%2Fironic-tempest-plugin~master~I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821","project":"openstack/ironic-tempest-plugin","branch":"master","topic":"patch-to-enforce_json_extension_works","attention_set":{},"removed_from_attention_set":{"36770":{"account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"last_update":"2024-05-23 06:56:36.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-03-28 12:08:27.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2024-05-15 15:14:12.000000000","reason":"\u003cGERRIT_ACCOUNT_11655\u003e replied on the change","reason_account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}},"hashtags":["ironic-week-prio"],"change_id":"I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821","subject":"Patch to enforce json extension works in existing API behaviour","status":"MERGED","created":"2024-03-21 19:13:45.000000000","updated":"2024-05-23 06:57:31.000000000","submitted":"2024-05-23 06:56:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":44,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"913926-patch-to-enforce_json_extension_works","meta_rev_id":"e6ff9ab5489029f10201bf4ed1bbfba2d3170f8b","_number":913926,"virtual_id_number":913926,"owner":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-23 06:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2024-05-14 22:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2024-05-15 15:14:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2024-05-15 15:14:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-21 21:18:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 00:07:43.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2024-03-28 20:25:15.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2024-03-29 19:53:50.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2024-05-14 22:51:12.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"96b9b4b54c1dffcc364ba2021a5ef2b8b7c7a5b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-03-21 19:13:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52221d1c0e4f618e0c540789c02f2e842331d621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 21:18:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6b817edbadf48b7af77edf8ddedf093\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a26516093de4bcba1f1efafb7100615 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2720f8cd2c04a0cb079d7b79664686e : SUCCESS in 2m 24s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/c5c57de4a987416780584d14351a3a44 : SUCCESS in 1h 02m 48s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/1429f37856ca464bb3b64b1304f5cae0 : SUCCESS in 1h 09m 03s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/a46630bb49d44d29a6886d968409411f : SUCCESS in 47m 41s\n- ironic-standalone-zed https://zuul.opendev.org/t/openstack/build/b158d0a4d24c450a8ae65febba805475 : FAILURE in 23m 07s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/87d5a6a64bce4f34843f10874d391bd7 : FAILURE in 29m 59s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/bfe3fb78c67845d2a68bcd4d27cd1002 : FAILURE in 28m 33s\n- ironic-tempest-functional-python3-2023.1 https://zuul.opendev.org/t/openstack/build/8ee5d8f0b1a84f4d8f8a28c2424019a0 : FAILURE in 36m 03s\n- ironic-tempest-functional-python3-zed https://zuul.opendev.org/t/openstack/build/648e09e58f76491d98a27b2173798c02 : FAILURE in 21m 55s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/9b955689347c4712a052d63bcac8bb74 : FAILURE in 20m 18s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/878f2a0e4a9e472fa714df270b4c2fae : FAILURE in 23m 01s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/8a75dbbaadef443997128832d3ce2cda : SUCCESS in 1h 07m 30s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/8c51b5154574437c894f56c99c3847e6 : SUCCESS in 56m 00s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/3062210ef6944a4f8af68f61aac15b69 : SUCCESS in 57m 01s\n- ironic-inspector-tempest-zed https://zuul.opendev.org/t/openstack/build/c5c9e95aff854197b5fb17c2825fa8bf : SUCCESS in 48m 21s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/08e25e29e8df4a6a8f14a13e15028770 : SUCCESS in 37m 01s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/893f2ee1396c4fea945d4b11cab841f8 : SUCCESS in 1h 02m 44s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/bcc61b83c7cb4cfd9c777354318278c5 : SUCCESS in 47m 11s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d67af90e3b12442cbfc8a421170176d7 : SUCCESS in 1h 21m 33s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/8a03c83325c94cbb99be87d7d015615c : SUCCESS in 49m 13s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/14224edc4cae466fadd2b8aaa209557c : SUCCESS in 52m 57s\n- ironic-standalone-redfish-zed https://zuul.opendev.org/t/openstack/build/1c0f9a1ecc324c64b15a5bcbca6f071f : FAILURE in 42m 00s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/621d3541a9d34bc9b047d39195c41efe : SUCCESS in 1h 15m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/07fab24330d34fac8e0b7b91f25903ce : SUCCESS in 1h 29m 15s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/adc949fc6ddf4dacb28e6a3d16601806 : SUCCESS in 1h 27m 08s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-zed https://zuul.opendev.org/t/openstack/build/82390e1611c64851aff7ace0ef2ca216 : SUCCESS in 1h 58m 29s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/8be6571f08754f7b99e883145ef3be36 : SUCCESS in 40m 17s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/6860e774e0634e39a8815c861b0b19b4 : SUCCESS in 48m 19s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/3e8d6f8545ff47eeb8ff8f082a764274 : SUCCESS in 45m 26s\n- ironic-inspector-tempest-discovery-zed https://zuul.opendev.org/t/openstack/build/c09c0808bd644ca39b742aea254f6f62 : SUCCESS in 51m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ede4f8ca60e207cc32027a6d56961c778212c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-03-22 13:53:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"82e42b12740a60f489cbd45c9761c2352d857bb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-03-22 15:35:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8cfc164b2262bbe0a29eb36b3414e867370ec124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 18:00:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93887882d20d4896823465ac7fb11499\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f29b7e7029324214beca12ff3973840b : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ac0d244c0a427cb8f2d445d27ddb73 : SUCCESS in 2m 57s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/0022bc5b545640fe860e5ba013ccf5cd : SUCCESS in 1h 03m 53s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/7303a4a419f04f3891c658deed6f441f : SUCCESS in 56m 18s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/a7f448c0c7454700b755c4898a5e545c : SUCCESS in 1h 14m 32s\n- ironic-standalone-zed https://zuul.opendev.org/t/openstack/build/7260b426ab8a41e0808d55c2e1e81811 : FAILURE in 56m 30s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0a54402af07d427ca091897cc5a48fa1 : SUCCESS in 34m 19s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/6709d8cfceeb4e198df634739ef57dd6 : SUCCESS in 26m 17s\n- ironic-tempest-functional-python3-2023.1 https://zuul.opendev.org/t/openstack/build/58bbfe717ff94bf882e5acba0bf215a4 : SUCCESS in 28m 27s\n- ironic-tempest-functional-python3-zed https://zuul.opendev.org/t/openstack/build/02a2b5c1bc484f70a9f65d8cbdd2e853 : SUCCESS in 32m 23s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/b4752f5f2f044380b17bb5357fbe23c0 : SUCCESS in 43m 23s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/e4bb4ca43ec6417fb6340e73ea1e607c : SUCCESS in 23m 16s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/1425b998e9804690a87c3a73c0e99a0f : SUCCESS in 47m 46s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/204ca0e7c1674fa38cb5b4a2f84d9596 : SUCCESS in 46m 25s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/7aef991eb8824fb8b8075387f03ecd5c : SUCCESS in 56m 32s\n- ironic-inspector-tempest-zed https://zuul.opendev.org/t/openstack/build/6b07cc4547ff40149761ec028d292aa3 : SUCCESS in 1h 04m 00s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/302f5ff03b7d4d32897092abf6a15dad : SUCCESS in 54m 30s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/f580c9ecb3aa4afbb6f311658b0ee242 : SUCCESS in 50m 27s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/8140d9cdf8b44928980eb3104dadcbf7 : SUCCESS in 50m 53s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/4dc97cbda9ec4291823bcf911ab62602 : SUCCESS in 1h 26m 12s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/b44307d2c0e340e4bc2e684f66b08960 : SUCCESS in 57m 10s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/24d9c4b09b9f48ada933b8730d213791 : SUCCESS in 47m 40s\n- ironic-standalone-redfish-zed https://zuul.opendev.org/t/openstack/build/66d98f0c89a44eac975057633200cf57 : FAILURE in 57m 27s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/e08bd445ee3146e09d28cd97b3cb1405 : SUCCESS in 1h 39m 38s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/9539be09cb944a858987fcf8323cd4ee : SUCCESS in 1h 26m 18s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/ca6368846e1a4587b2923aea06409535 : SUCCESS in 1h 53m 38s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-zed https://zuul.opendev.org/t/openstack/build/cace682fa7dd4fe4ad25fb18d04e9d02 : SUCCESS in 1h 47m 39s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/f78918cc727e4d6b88f1b9bccf95451b : SUCCESS in 58m 05s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/ef875919ef6449f0aee72c6739e22bdb : SUCCESS in 46m 04s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/06b12c5d99994005bc69e54a9e14e34a : SUCCESS in 35m 27s\n- ironic-inspector-tempest-discovery-zed https://zuul.opendev.org/t/openstack/build/e09913f497514a80a950f91240c4dc1e : SUCCESS in 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a5f0396de5619ae06fbb6a0c37b915a0efffbc41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-03-25 20:23:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"89722c213a2a56fe8b9664a18c5c620b931d7e4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 22:35:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d841380b592415580e112ad655e9053\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d248437c0eb74b69b649f0faf2841c6e : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63feaaaf1e984ee99af0a7d401485f5c : SUCCESS in 2m 42s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/b8b96e4f9c104cbeb2245263ca8fe430 : SUCCESS in 1h 08m 12s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/19736249cb964e0183f0d7c62ba50200 : SUCCESS in 1h 28m 04s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/b32c30651d334ab49e3c966d8233c3a1 : SUCCESS in 47m 45s\n- ironic-standalone-zed https://zuul.opendev.org/t/openstack/build/7387d769aa47422faf153ba1883ba675 : FAILURE in 50m 00s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ab2754eba6f7419e949c9e76c02fb00b : FAILURE in 31m 05s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/e102cf9f0d5243c88bf48883e387cf39 : FAILURE in 32m 26s\n- ironic-tempest-functional-python3-2023.1 https://zuul.opendev.org/t/openstack/build/6057b5ad93044ac382d88766a74e334c : FAILURE in 32m 06s\n- ironic-tempest-functional-python3-zed https://zuul.opendev.org/t/openstack/build/fb30e2a4d59043118c8c50147949d7d5 : FAILURE in 33m 41s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/829cc68c66ab414f98c76259985bcf69 : FAILURE in 33m 45s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/5cf287c2fb1641ddb35ddcfc4e5f5b06 : FAILURE in 32m 15s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/54dbe7a4ed934b1f9f6c5003536940ca : SUCCESS in 1h 02m 13s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/e09d8615abc94c19b81a97f9fe1d07a3 : SUCCESS in 58m 19s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/f3f25611c6974443a2791228a010996e : SUCCESS in 1h 01m 55s\n- ironic-inspector-tempest-zed https://zuul.opendev.org/t/openstack/build/c89cf8fb5e704d6c9e649be30652d655 : SUCCESS in 58m 58s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/4dcc2d47cde246c4a25102c5a24b6683 : SUCCESS in 52m 56s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/cef85e48517648b1a8d8250cdd172c29 : SUCCESS in 51m 50s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/f539011aefdb409c9bf9b473c15e4e36 : SUCCESS in 34m 28s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/5fbb43489bb647b693b8a781af1ab493 : SUCCESS in 1h 04m 47s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/46df77850dc741e79b0b95c84ef928bb : SUCCESS in 1h 15m 13s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/d5308c5a5c364586a852e252b153bdb3 : SUCCESS in 1h 02m 10s\n- ironic-standalone-redfish-zed https://zuul.opendev.org/t/openstack/build/4fe51d1869d04b2ebfd3bb041b90c224 : FAILURE in 54m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/b8eaf346f7cf4250bc18e8d5b1632a58 : SUCCESS in 1h 22m 59s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/8e20a46b0e8f47788f7824c8c42848b9 : SUCCESS in 1h 47m 30s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/0b6a2a1a41834336bfb239fd6d0f337c : SUCCESS in 1h 50m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-zed https://zuul.opendev.org/t/openstack/build/cd00aaa9e3bc4cf8bd0e3ee156eb5d8e : SUCCESS in 1h 35m 54s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/51d1b51a033f44388cf543f37beb03f8 : SUCCESS in 46m 47s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/c8fa95455a46496a8c1aa9791f9ad87d : SUCCESS in 36m 34s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/a8a02ea4d9fd45d7a873785fcb1ce367 : SUCCESS in 34m 48s\n- ironic-inspector-tempest-discovery-zed https://zuul.opendev.org/t/openstack/build/9ed23c51001c48a69ef27105981b1dbf : SUCCESS in 39m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"175253d8796bbd625bfc09f3e224e17af3dec1d6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-03-26 00:07:43.000000000","message":"Patch Set 4:\n\n(9 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a502cfe7f40c0f5e79f86b3437f271022b37b110","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-03-26 13:45:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"25669cdb8e727117210f895491fe7d77ad387404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 15:44:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12a1cabbec044300aa147ff4ba322860\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca0bb1f01e541d699f9e493f3420ea2 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f02c300936d946daa0afd1e6c47a8a05 : SUCCESS in 2m 31s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/204b1649eed9488484130c068b684f46 : SUCCESS in 1h 04m 11s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/3127991afcd54026b06e13759f9b7647 : SUCCESS in 1h 12m 08s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/d237d3db7c46416aa63358f190167f60 : SUCCESS in 1h 19m 57s\n- ironic-standalone-zed https://zuul.opendev.org/t/openstack/build/b6cdae60276746fa8c9ba65bca1444b5 : FAILURE in 59m 59s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c3f4cf09ec7f4865a0339aa698a37972 : FAILURE in 33m 59s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/17f720d5c6994c1087602ce0ca568a03 : FAILURE in 31m 03s\n- ironic-tempest-functional-python3-2023.1 https://zuul.opendev.org/t/openstack/build/c65787a1d941451a8d09e58e3c4f5e12 : FAILURE in 32m 06s\n- ironic-tempest-functional-python3-zed https://zuul.opendev.org/t/openstack/build/bfdb4492b9a147d18cbb859f188d62c0 : FAILURE in 30m 28s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/0e22208228e240e3818e17a9ccccc16a : FAILURE in 34m 13s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/4854bc0b5f384cc5b283fa1aaa201044 : FAILURE in 24m 30s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/c57ac74972de44b8b5e0bf65203d4b2f : SUCCESS in 49m 47s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/d796071891164ae28eb5071ab322c128 : FAILURE in 45m 20s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/917270f85f1a45189fd5c4cc98df7a0b : SUCCESS in 50m 45s\n- ironic-inspector-tempest-zed https://zuul.opendev.org/t/openstack/build/a90b8ee5ecf9493781568c69fe9a7b8f : SUCCESS in 52m 02s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/19923408ef4b44cab90c4bdb26f74670 : SUCCESS in 53m 11s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/9ad6f6b11a854a0a8b5bc61f0b941eb7 : SUCCESS in 42m 50s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/23863a127e924b4696fedba361188492 : SUCCESS in 38m 15s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/358fbf4285674f409ac1a3b79615307c : SUCCESS in 1h 11m 21s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/721ba8863adc4f9d803f078a6a201395 : SUCCESS in 1h 01m 30s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/97e358e9803a4c4bb19c9cbe78e947a9 : SUCCESS in 1h 17m 23s\n- ironic-standalone-redfish-zed https://zuul.opendev.org/t/openstack/build/7be3e40f633f420bbfe68e85598172f9 : FAILURE in 1h 13m 25s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/13b8e02fd2f446fc99fd4287dea73e0c : SUCCESS in 1h 27m 19s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/d50f411469da415f88dd19be22d3b0e6 : SUCCESS in 1h 52m 48s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/1bac0ecf8a864be792754521de146d1b : SUCCESS in 1h 45m 02s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-zed https://zuul.opendev.org/t/openstack/build/9b360650c52c4818ba56c7a0175c9e9c : SUCCESS in 1h 33m 04s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/6fa85af5c1ef4bf5af1a8fac4d652e05 : SUCCESS in 42m 50s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/86e8774edb024f758c2b15f130c105b3 : SUCCESS in 45m 36s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/a790a791ec314625bb10ab550b57eb99 : SUCCESS in 47m 30s\n- ironic-inspector-tempest-discovery-zed https://zuul.opendev.org/t/openstack/build/4b5bc661f65044c0800bd4ed92846d33 : SUCCESS in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"49391097884bff1d01b207efef63167986d4e23c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-03-27 17:14:17.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c46c44ebe42f444101fc99ce306ff3129f4caa13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 17:14:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d04b13abbf94fbc91c8e39793d8fcbb\n\n\nWarning:\n  Dependency cycle detected and project openstack/ironic doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":6},{"id":"38e48020c72d9e059e5fc1250328137dc5f8ab2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-03-28 12:06:59.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"8852c9de8f97f0983618582439c1c89ec5695c07","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-03-28 12:08:27.000000000","message":"Hashtag added: ironic-week-prior","accounts_in_message":[],"_revision_number":7},{"id":"2d5727f438d0ca71eed0eb033724372cfc9c9989","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-03-28 12:09:37.000000000","message":"Hashtag removed: ironic-week-prior","accounts_in_message":[],"_revision_number":7},{"id":"66aac5b895fc8bbe7170e0b338c2f8f5a12b44d7","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-03-28 12:09:44.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":7},{"id":"6948b20477d6a8152dcd352c736a40faba07ac73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 14:21:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3d35e6667654ac5813f9323e81ecf4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e064307f0834fda87feac31df1fd0e2 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd135c525f14f9a8af965d0bd0da88c : SUCCESS in 4m 12s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1cef258d591747109cb230f5c9251f0e : SUCCESS in 1h 14m 51s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/7d932510250a4328b28d16ce8b4e5272 : SUCCESS in 55m 03s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/6e74e1abd9194e0d81a9327f577d4203 : FAILURE in 1h 12m 12s\n- ironic-standalone-zed https://zuul.opendev.org/t/openstack/build/ac063d47adf34f318acab1d657530ad5 : FAILURE in 1h 01m 02s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/73382d3caa5b4d20b5cf8812186b9345 : FAILURE in 38m 25s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/626bad4dcf424b24a63a60eabd8b49dc : FAILURE in 35m 13s\n- ironic-tempest-functional-python3-2023.1 https://zuul.opendev.org/t/openstack/build/3319b571c638440e829ff58ca72fb240 : FAILURE in 30m 46s\n- ironic-tempest-functional-python3-zed https://zuul.opendev.org/t/openstack/build/79725f5672ec4bf494289b21b532eda4 : FAILURE in 34m 36s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/3e928e0629b5411fad6a8201e8fc9bf2 : FAILURE in 32m 18s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/4e43dfab62384dda8b4a77dcd66f4780 : FAILURE in 31m 19s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/850b4d40c6424c43a313497dee231bf4 : SUCCESS in 1h 05m 09s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/56dab370462e4011a9eb72fac323786f : SUCCESS in 59m 30s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/6f61c6c2a1214d4b806ff582ca37678a : SUCCESS in 57m 42s\n- ironic-inspector-tempest-zed https://zuul.opendev.org/t/openstack/build/8fe8277e8e0a4b3abfc4e3d0c19d08c1 : SUCCESS in 1h 00m 50s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/d05d2b55c8fc40d3ab62b896990c1daa : SUCCESS in 58m 38s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/518da8b83d4b4a40aaf9db88671c6ff3 : SUCCESS in 46m 36s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/88365bb03c2141c4b45060e1847fffee : SUCCESS in 53m 27s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/064ae680142641dbaf0c5d772db77672 : SUCCESS in 1h 26m 56s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/3da088df27de4cb58df33042a0d8099c : SUCCESS in 59m 17s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/4e2225506ca84bfcaeeee187010ce5e7 : SUCCESS in 1h 04m 33s\n- ironic-standalone-redfish-zed https://zuul.opendev.org/t/openstack/build/3d718f7932fb4dfb8962a5bb94f61f30 : FAILURE in 1h 02m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/827c36a29d014e66af928e3d877b334a : SUCCESS in 1h 59m 40s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/beceaddeb21848f3aecc7ce37d2d6a1c : SUCCESS in 1h 58m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/a86f7bf5c0dc494c9b2cc7271de5927b : SUCCESS in 1h 35m 22s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-zed https://zuul.opendev.org/t/openstack/build/27aa124c0905469195b0f9d0316c344c : SUCCESS in 1h 52m 42s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/5ca90349bbbb4d4c8095e92bccf9159d : SUCCESS in 59m 00s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/f592f37c429c422f83b77e5f6bfc474b : SUCCESS in 49m 32s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/ec9773bc36534bafbce2f78b9d9b8ca4 : SUCCESS in 49m 36s\n- ironic-inspector-tempest-discovery-zed https://zuul.opendev.org/t/openstack/build/4add958bc10f49b8b67bc21878bb9b48 : SUCCESS in 47m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6b1a5d1e01f1b12c614562581006e94b68f7accb","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-03-28 20:25:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ddf758aefbe279f73dd8f2782564a47e43d02779","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"real_author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-03-29 16:56:30.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_36770\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}],"_revision_number":8},{"id":"b5e9cf5f8a8fbe03b0fa45b77b8497e476bc18c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 19:08:39.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11f14a784c7b4772a902f8255cfa17af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf2a13fb25dc46aa9452c432357aa066 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea3e8196aa04d868e19441f4bfed0ec : SUCCESS in 2m 49s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1f5c306b99c94db9b599b1e867cd354f : SUCCESS in 1h 02m 40s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/1d5efb7f9c384ab6a489fb863e2fef52 : SUCCESS in 1h 12m 27s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/4e12ddd89ab74d6599ae5e45812e3c5d : SUCCESS in 52m 21s\n- ironic-standalone-zed https://zuul.opendev.org/t/openstack/build/d1419d4552c347c5bca4d1e18b169cca : SUCCESS in 1h 03m 00s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/3544f2a42de34690a561b13d5ada8cfc : FAILURE in 34m 04s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/d7e4f0b366b140f987900dc9cc28b215 : FAILURE in 32m 20s\n- ironic-tempest-functional-python3-2023.1 https://zuul.opendev.org/t/openstack/build/b71b365f678947929feb6ef3f207a2f4 : FAILURE in 28m 40s\n- ironic-tempest-functional-python3-zed https://zuul.opendev.org/t/openstack/build/111e854902e64e2087803bb239a70684 : FAILURE in 20m 56s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/c6e63e841dcc4909b05e6a273779fad2 : FAILURE in 33m 21s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/4375e335fa404be9991deeea0f219abc : FAILURE in 31m 46s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3687349b2b51474bb0db7631628fe722 : SUCCESS in 1h 09m 11s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/3524ad46810143fd9b7722408e46b302 : SUCCESS in 1h 00m 52s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/e10a65b10e264479a6c39db54a3ae9de : SUCCESS in 49m 25s\n- ironic-inspector-tempest-zed https://zuul.opendev.org/t/openstack/build/64ed9e8b26034bccb15f37013dda6a6d : SUCCESS in 45m 11s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/ef418618cadb4845b80cce5db56fbea8 : SUCCESS in 1h 00m 18s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/4a5b161bff734012829ba161b30fe351 : SUCCESS in 47m 53s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/442f15d4728d4f378098708fab6d243d : SUCCESS in 49m 50s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/677f288c080640538c931e9a3b397b26 : SUCCESS in 1h 03m 38s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/fcb8b61ddb8d4b21a7dc8d2887de1a1f : SUCCESS in 56m 37s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/cd1ba11f137040fda6e3c323ead6b728 : FAILURE in 55m 54s\n- ironic-standalone-redfish-zed https://zuul.opendev.org/t/openstack/build/cfe733476f6740a9bc6d36f3e2e9fd42 : SUCCESS in 1h 09m 51s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/48a485ad39da418da2aa699306248132 : SUCCESS in 1h 47m 24s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/ed032a0f36f84b919f70193d6dd89174 : SUCCESS in 1h 30m 57s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/458ade8061244678b8900e0a163688d4 : SUCCESS in 2h 00m 08s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-zed https://zuul.opendev.org/t/openstack/build/d7576f33d5484ce7b2b503259c96d223 : SUCCESS in 1h 39m 02s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/8d92fb7b90fa401e9beabdbec25f5df7 : SUCCESS in 38m 48s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/d06f28196cba4d548740178224f61b09 : SUCCESS in 34m 44s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/8f23ddab7c974e83b68e23783f0575c1 : SUCCESS in 43m 11s\n- ironic-inspector-tempest-discovery-zed https://zuul.opendev.org/t/openstack/build/8d072af362db4e0b924e78ccd0599b0e : SUCCESS in 48m 23s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d9a17ef3975760172fc067ec3547a61192aed37c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-03-29 19:53:50.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"43316176b92de60528d5caa0057c7a08f3885bf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-03-30 09:15:05.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"45ac35f284f166af26eb401bdd7fb701c069528e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 11:26:58.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/920419d202e5434d97043a4f857b138c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a660671e7ee44438a64463ac8bee0537 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b257b56fe5940bba8904137309b1814 : SUCCESS in 3m 18s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/f75235e671e94b4a8366056bb1064030 : SUCCESS in 1h 16m 12s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/cf6c1f41b5d14c1c9b55acc1e24a4d03 : SUCCESS in 1h 22m 33s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/19d771b93bf34bd8a3d349d4f901ac70 : SUCCESS in 45m 54s\n- ironic-standalone-zed https://zuul.opendev.org/t/openstack/build/484a0fd8d5c043609a95ad18b87a3627 : SUCCESS in 1h 36m 51s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/949d14ef40c24b22b9da8d693370a2a1 : FAILURE in 21m 37s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/5e78554c7fbf4cefa35a3aef52dc21dd : FAILURE in 32m 06s\n- ironic-tempest-functional-python3-2023.1 https://zuul.opendev.org/t/openstack/build/95c293cb610f48f993d2d0a3b1f7816e : FAILURE in 24m 03s\n- ironic-tempest-functional-python3-zed https://zuul.opendev.org/t/openstack/build/ee261f54e91644b2b978a55c643e18ac : FAILURE in 33m 43s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/83fe6c96201544fdb3da92ed4e46674a : FAILURE in 22m 05s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/0a0f7dd0f8df43aa974d135ab05467bf : FAILURE in 34m 00s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/a671dcc75bf349e9bdd13139243dd2bf : SUCCESS in 56m 56s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/53c9fdf0ddd14fda8039ba435e8f3d2d : SUCCESS in 43m 32s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/a5254d434bef4f1f827afedff4b67fc1 : SUCCESS in 49m 46s\n- ironic-inspector-tempest-zed https://zuul.opendev.org/t/openstack/build/85a6c4f54b19436783dfc9717da531f7 : SUCCESS in 50m 03s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/a4c63ab67fc444b8a00e4be0d9f593ca : SUCCESS in 34m 57s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/a3aefaae1a5d4a708db1081a6d9b5c20 : SUCCESS in 34m 05s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/e831f39583264f44963232e00ef36aa6 : SUCCESS in 36m 40s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/26475e61e125428e9b0e60cebcd88b86 : SUCCESS in 1h 05m 56s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/c96e27b5d4ab40ec93d28469a95a7c71 : SUCCESS in 1h 03m 43s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/31c3bc3986d64ee5ae77c8f4d560a76c : SUCCESS in 1h 20m 55s\n- ironic-standalone-redfish-zed https://zuul.opendev.org/t/openstack/build/be9cb8945231465898253cab987f0d85 : SUCCESS in 1h 09m 47s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d15d2a352e5a49acb9b9c2a079159e43 : SUCCESS in 1h 36m 47s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/ad978d48562a4422a9d0377041da611f : SUCCESS in 1h 47m 27s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/107a972d5af44cafaa3120d7d795c9e2 : SUCCESS in 1h 56m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-zed https://zuul.opendev.org/t/openstack/build/e59d22b5b6b04d81841ac1c14ea1d2d5 : SUCCESS in 1h 33m 42s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/3e4aac998e9b47d58ad732c7a30cbd06 : SUCCESS in 41m 44s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/2f8cbb945c6f4f68947b8f84b86f2c68 : SUCCESS in 45m 19s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/5d3b9d528d80401e9c21e3e6e26ce23d : SUCCESS in 46m 43s\n- ironic-inspector-tempest-discovery-zed https://zuul.opendev.org/t/openstack/build/9fb697c21daf4646a82a2f5399d91b61 : SUCCESS in 46m 25s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c2a831c5192ea98c55d1364f2f65dbcb880da936","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-04-03 20:34:19.000000000","message":"Patch Set 9: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"353e084a910e53d4b0cc636d1a6c5ef6d9215d26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-04-05 14:13:26.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"6522cf105fbed9024e996e049028f7d7d36342ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 17:05:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d65fafb48fc4a14bbb3d96822f9a5dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dcf3235e9994df38ea52cb942a48405 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6a468c1be54eb1b77ddc31a0f0af88 : SUCCESS in 2m 34s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/321d50543baa4909879ee370300a7b15 : SUCCESS in 1h 05m 18s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/735ecdc7e1f345daa08b8be2d1098564 : SUCCESS in 1h 19m 03s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/5680a968112e417181cc6d538b87808f : SUCCESS in 1h 00m 05s\n- ironic-standalone-zed https://zuul.opendev.org/t/openstack/build/f04115abba0b413a872725e517cc4a86 : SUCCESS in 1h 04m 59s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/306f5b1657b4467889a7d14e102ee0e2 : FAILURE in 32m 03s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/8d745d63ba4e4f26a7219723eba7afa5 : FAILURE in 21m 41s\n- ironic-tempest-functional-python3-2023.1 https://zuul.opendev.org/t/openstack/build/caff21e53cd5407492e47b53c4f705bc : FAILURE in 24m 47s\n- ironic-tempest-functional-python3-zed https://zuul.opendev.org/t/openstack/build/5ca835bc88f345819a127d407b874756 : FAILURE in 21m 08s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/1b1098d87fb24dc4b997b53f17896e4c : FAILURE in 36m 04s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/5aa784dcca1c4d2daf1a5d0077659b23 : FAILURE in 32m 47s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/fb3a8ff62f944db1a2d5b88473785a90 : SUCCESS in 59m 35s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/231beabccd1e41a688374801d9ee658e : SUCCESS in 45m 49s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/7db0045c14074e289f33edec048bfa91 : SUCCESS in 1h 07m 10s\n- ironic-inspector-tempest-zed https://zuul.opendev.org/t/openstack/build/ec8aaa9102da4968859f2ec5439507e3 : SUCCESS in 1h 01m 52s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/22e1a1a7ce8446e99db5190acc8c29ae : SUCCESS in 53m 09s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/c911bfc46e5b40da8db069edaeff4662 : SUCCESS in 33m 14s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/45ccf31055a44715ae4923d9c31f611c : SUCCESS in 45m 02s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/0ca29ce51af347b08b6469a940018a56 : SUCCESS in 55m 16s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/4383ae01d3564175a64bec0c5ed2cbeb : SUCCESS in 45m 21s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/6a01154a4bd243f0930e6a020b3967a0 : SUCCESS in 57m 33s\n- ironic-standalone-redfish-zed https://zuul.opendev.org/t/openstack/build/675a7eb17b1c4a21a615fa41f91dc38e : SUCCESS in 47m 56s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/82384395cf4946e8b42bef8fb9cbd8fe : SUCCESS in 1h 51m 15s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/59a6871ce8d94d34a7894eca41576d02 : SUCCESS in 2h 10m 14s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/4c5585fbf6674fb085a8fe79730f06fd : FAILURE in 13m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-zed https://zuul.opendev.org/t/openstack/build/f6ff689332784a7ab913754323402674 : SUCCESS in 1h 12m 39s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/6f902c5494114ce8b8b6c9af13085fec : SUCCESS in 48m 51s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/c412e5e9a4254e79afaabe83ba3af61f : SUCCESS in 35m 33s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/07acf6def73e4c2cb64ca07bf7c6b9ef : SUCCESS in 39m 55s\n- ironic-inspector-tempest-discovery-zed https://zuul.opendev.org/t/openstack/build/9403341629ba4b2e8f128be959cbbdef : SUCCESS in 37m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6d2a27ee7a36d47ec6e17f440cfc74cd87850f80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-04-19 16:48:38.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"08723c3df0ecc24c9ca83ce2204a35c058d08922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 18:49:46.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a32b159d23f43418bf4d6c01487ae56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147dfd7b2670405698ce93df75817f58 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/309830ba0cb34f07b5c66cf9bc1b4a53 : SUCCESS in 3m 43s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/cc47870b1fb344de8fa1874c97ea04ff : SUCCESS in 1h 00m 39s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/8e430b7b502c426bbe2f128aa219a661 : SUCCESS in 45m 14s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/ed0ce55565404ae58bfedbec7f36e27f : SUCCESS in 44m 52s\n- ironic-standalone-zed https://zuul.opendev.org/t/openstack/build/dfc11b10b7c541f28f00dc4d377b0e61 : SUCCESS in 43m 15s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/f7c7b75e81da4d7c90df88bf037da528 : FAILURE in 33m 31s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/10007cc399584fad97a177701600f0af : FAILURE in 33m 39s\n- ironic-tempest-functional-python3-2023.1 https://zuul.opendev.org/t/openstack/build/410b181a2f694cc88447d3f3d208bf0a : FAILURE in 32m 45s\n- ironic-tempest-functional-python3-zed https://zuul.opendev.org/t/openstack/build/ac80410e0b4147b996a6558957710631 : FAILURE in 21m 49s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/ebe99948b3b041598a5cdd4f9fb6e8bb : FAILURE in 21m 53s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/2c56caaf1fcb4dc8a9fca6b929ccd134 : FAILURE in 26m 17s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/43347a1b7f96491e97359d146547a24c : SUCCESS in 48m 10s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/191f38f9a01446209b85290773a0dfc7 : SUCCESS in 54m 22s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/f8548810c4644813a61cd0e2f232b98c : SUCCESS in 45m 41s\n- ironic-inspector-tempest-zed https://zuul.opendev.org/t/openstack/build/240b65b62c994a178094fa87daab8c5e : SUCCESS in 54m 47s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/242d80c9413745d1944841134ddbe98a : SUCCESS in 36m 15s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/a76cf12e4be14cac946b24098814e88e : SUCCESS in 40m 48s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/317b5cacde7147bc9fd4e7abe5fc7482 : SUCCESS in 34m 55s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/e1d1cefe05f048b68a6b2890fe09019c : SUCCESS in 1h 54m 52s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/6b0f7543de1b4d3ebb0d25b013ed3a4f : SUCCESS in 1h 16m 03s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/627dd428bad943e490da799aa9a9d805 : SUCCESS in 1h 00m 25s\n- ironic-standalone-redfish-zed https://zuul.opendev.org/t/openstack/build/4be5818a590d4bfe9aef9205dc78e019 : SUCCESS in 51m 45s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/e47600761e334ceb84275fd3cd9f65df : SUCCESS in 1h 09m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/1009945d18824200ac77e5b81efb4ace : SUCCESS in 1h 31m 53s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/6b3dca9e7e7b48ccb1fd9b5bb34fe16c : SUCCESS in 1h 53m 24s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-zed https://zuul.opendev.org/t/openstack/build/2351c25bd1bd48f3b944d7848507d362 : SUCCESS in 1h 39m 28s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/2b24e68e2fc14cd19f3a6c923e7dffa3 : SUCCESS in 34m 21s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/cc91795137564cbaad40b238f53aaa51 : SUCCESS in 34m 19s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/16bd9dfdb15e4a7ca5e31babbf24ee5b : SUCCESS in 40m 04s\n- ironic-inspector-tempest-discovery-zed https://zuul.opendev.org/t/openstack/build/661950c938364910b1b248f53650b425 : SUCCESS in 43m 10s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c0f3969f410a055f0ad909a94d34db7e7ba0c710","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-04-22 11:18:36.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"bce45baa3061e1a0565fe20114912d14fd0c18bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 13:26:46.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc07ea5a82114891b0074cc76eb875aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/113269291cd240adaeac631025634e5c : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0129e96ac18d4891b85ced13906671eb : SUCCESS in 2m 31s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/93525375d2a7490f900a04e711bd7604 : SUCCESS in 1h 04m 29s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/8a3f48a6d1bf45ed81f9c5b4af95e54b : SUCCESS in 42m 41s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/0a9052aeb4504048a2eb2cbd5b22bcc0 : SUCCESS in 57m 57s\n- ironic-standalone-zed https://zuul.opendev.org/t/openstack/build/8ba1d345f7cd49e6afd35cd3264019f5 : FAILURE in 54m 30s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/47ffe65994b446fa82cd3ea95320dd8a : FAILURE in 23m 49s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/4595ad99f93147e19144869ef6a206b8 : FAILURE in 36m 43s\n- ironic-tempest-functional-python3-2023.1 https://zuul.opendev.org/t/openstack/build/287b6c562cc84370b081243d60e65106 : FAILURE in 33m 53s\n- ironic-tempest-functional-python3-zed https://zuul.opendev.org/t/openstack/build/33cf2981c80f479eac1abe72132182f9 : FAILURE in 22m 59s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/21141db9be674fb1a7eed3644ccd3639 : FAILURE in 21m 18s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/3c4511b9717c477fa89b67e05e9fd310 : POST_FAILURE in 32m 21s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/d6ee291032564e8a96cfa440f6f685ac : SUCCESS in 57m 00s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/7ddc91b956814b7abb6f90c3a4e5e4c0 : SUCCESS in 1h 01m 42s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/0be07db3b9d94dd8a05a6aa1048456e9 : SUCCESS in 1h 00m 26s\n- ironic-inspector-tempest-zed https://zuul.opendev.org/t/openstack/build/5241ad5b5b77451e8ff989f799a362ff : SUCCESS in 48m 41s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/599fa49f9f1242889df0bc22fc38c942 : SUCCESS in 34m 44s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/396d05c619394d9fb54a6bebe8000837 : SUCCESS in 49m 40s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/9ed653a2c3cc4096b78ca813d11acc96 : FAILURE in 32m 19s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/5af403747e3b4af7841bb5df6eea89b4 : SUCCESS in 1h 25m 29s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/c2979cb982d5411dbf035e787af4f379 : SUCCESS in 1h 03m 15s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/b30bfa691830444fbd490d1bfff5cd61 : SUCCESS in 43m 38s\n- ironic-standalone-redfish-zed https://zuul.opendev.org/t/openstack/build/4c22eeb7812544e6914c12413e86bfeb : SUCCESS in 46m 34s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/fbaaf34ca6154606b424f15d1b9ad3bf : SUCCESS in 1h 38m 52s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/80918660c9a945d0a3c3a180548bcfa0 : SUCCESS in 1h 33m 25s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/e6e34fb96d8b4500b7aa6e83e525dfc4 : SUCCESS in 2h 02m 38s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-zed https://zuul.opendev.org/t/openstack/build/9b99dd3fa2504f089c201bb1966720cb : SUCCESS in 1h 35m 25s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/e116ad1c9b4047439e7ab7e763730109 : SUCCESS in 32m 25s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/e17afada0d5a4b9b9d605d28f8aba9c0 : FAILURE in 39m 26s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/31c9722fea1a4a9fa4c2d7bc0b38bf4b : SUCCESS in 39m 03s\n- ironic-inspector-tempest-discovery-zed https://zuul.opendev.org/t/openstack/build/113b99bc8efb4640871e2e4a932be25d : SUCCESS in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d6eb1187510484bbcc54ac2c6554249c5c9475ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-05-07 12:31:32.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"6af1695944fd1f8e98020a8d8f00395f9ec7ba89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-05-07 14:22:41.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"5203a6ab67f0aeb2832611a74651d684c4f36c80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-05-07 15:49:16.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"66944919e4cd5876a23e2ec99be3b98a30ea9825","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-05-07 16:13:25.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a314e8a904515a0440ace439195902b8f490e125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 18:21:05.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0d82a1cd65f40b5bde7a23029480785\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5669ec4a6d844878a685e3d6fe5cd12 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e123ab99bffb4c48b13a21f079540a97 : SUCCESS in 3m 03s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/a89ecbe841504eef8403d786f19ce326 : SUCCESS in 49m 24s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/b569a68450134af38bb77c4313ddb79a : SUCCESS in 50m 18s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/521597da6f184855b9bdc26bc0857250 : SUCCESS in 51m 10s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/6e5cd3e717de483c93bf460519446658 : SUCCESS in 1h 02m 03s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/4a98e8cc3fc9407eacaf4aa35b8299e6 : SUCCESS in 24m 47s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/3f394b734f6743dc898e171123c8c463 : SUCCESS in 26m 20s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/8aff39c620254aafb22004b644dcf803 : SUCCESS in 33m 17s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/9bc0d51cdbfd4407bafb08f7784694a6 : SUCCESS in 35m 53s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/821d61c7ac46402a814a47d60be13047 : SUCCESS in 21m 00s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/b4a86bf0562d4a9d925174903a493b48 : SUCCESS in 23m 01s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/2441ae3989ea4e46baaef72542f46751 : SUCCESS in 44m 29s\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/a4bff9ee3583443e8a46ef919c347f85 : SUCCESS in 59m 08s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/bb38bebe1eb94b9f976f3381b16504b0 : SUCCESS in 57m 57s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/88ab9a20359441f58c1568f5327daa33 : SUCCESS in 44m 51s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/9ffbdc4827bb41c6999d8d4e71f1659f : SUCCESS in 45m 13s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/6a54d7ed2ed54bd39392a3ab4c9d94f1 : SUCCESS in 33m 21s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/6260a3535b8e461887b7f0aee997675f : SUCCESS in 51m 24s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/50d75800f46c4ee5afbfc613b7b2e898 : SUCCESS in 50m 26s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/88fce496d27249a483d8eade7fce4eda : SUCCESS in 1h 01m 42s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/727a85839d7e420eb9a100bc0c0d963d : SUCCESS in 1h 07m 45s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/2eb815937fba4b0aa8ee5ed57c8f73f0 : SUCCESS in 1h 24m 35s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/331c01d930974e5c8281b09228c2e8f9 : SUCCESS in 1h 03m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/9f3bb91d859c43aa99a2e01d13eb6a9c : SUCCESS in 1h 15m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.1 https://zuul.opendev.org/t/openstack/build/78f94daf8a814875aed5b6a66b062afe : SUCCESS in 1h 17m 21s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/b0312f31ebab450097c6505a0bc0f9c1 : SUCCESS in 1h 22m 44s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/73d97dcd5e504e0097067f2a4644dc2c : SUCCESS in 1h 52m 45s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/76f6ba68ee2d40fe98a69ca113077a65 : SUCCESS in 37m 41s\n- ironic-inspector-tempest-discovery-2024.1 https://zuul.opendev.org/t/openstack/build/d109c2475fe64ddbb41a109c496fb382 : SUCCESS in 39m 26s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/c0b7d3e005694ff0a8ee450f3b30f06d : SUCCESS in 35m 58s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/46d761df750b4d629cbeb56e27e5f041 : SUCCESS in 50m 03s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"89f80f9183d8f85770546ae31dbbda18230822fb","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-05-07 19:10:00.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a7033b4e0ec2fa77760079cf89e9c545b7c107dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-05-12 05:11:26.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"2577484d36b5670f86b195c207dcd806ba116759","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-05-12 05:19:48.000000000","message":"Uploaded patch set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"cade47b31a4331b49b51d738d3c062a7ee214438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-12 07:04:01.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9110c69c7f284a758ae515c04b3421f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4abaac6567784701b43c9f786f7b4e50 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e033d25182024a0fa49dcc353290fb06 : SUCCESS in 2m 41s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1aa78518123945dfb4cd34554af84116 : SUCCESS in 1h 18m 15s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/3f1c6e67078c4bffaf2fdbdb403bcdff : SUCCESS in 1h 03m 58s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/3dbcdbbc049448bfa115475db4911552 : SUCCESS in 1h 11m 08s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/87eb277354b747d384b759cd9d45d4ad : SUCCESS in 49m 13s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c0192e4d5cb04a119c1dded0586ddf73 : SUCCESS in 35m 58s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/cc8cf6cd1fb546db8ad0c430ec8129cd : SUCCESS in 33m 16s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/c680dea87a3746769bedd9927343a03d : SUCCESS in 35m 05s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/08005356c4884a9bac423643c42740d5 : SUCCESS in 23m 05s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/964e4c9471d3457ba936f705638e46bf : SUCCESS in 30m 48s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/6a67d9ff719b48ef838efb756ac9c693 : SUCCESS in 32m 23s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/5664084a500041f6ab03b6010575e441 : SUCCESS in 37m 51s\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/4565b7e40531406981872918359d9af9 : SUCCESS in 56m 00s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/25a3ee943b694118b327452f4f764944 : SUCCESS in 38m 56s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/2f6a7a8802d04558859fae8a4a4a49aa : SUCCESS in 40m 20s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/e227f1a78a9f4391ab4065ef2f365084 : SUCCESS in 27m 51s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/730981ae228c4edcae9e5f199c634886 : SUCCESS in 28m 15s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/2ace7a114d40404586f40f7aec88508a : SUCCESS in 29m 04s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/df275169e91b460d9f25d728cbf879c0 : SUCCESS in 29m 17s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d80a01620c9c4b28a1dc5f4d549e8db9 : SUCCESS in 1h 29m 13s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/c12d28aa28dc4f7bb69ce474dfa2fa82 : SUCCESS in 1h 04m 00s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/2297fbd6579c45689a1013429fe2ece7 : SUCCESS in 1h 13m 26s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/5fea1aac096b4235b07475c005d9ce75 : SUCCESS in 1h 08m 52s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/cf306596e6c54bdeb6fe583e93a04f17 : SUCCESS in 1h 32m 33s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.1 https://zuul.opendev.org/t/openstack/build/ddbd9b53358f46a48a3f0a90e63ee931 : SUCCESS in 1h 35m 00s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/8bbcbd9b0f8a48e2abf1a6e5ff6dbecb : SUCCESS in 1h 25m 39s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/f9400dd2b4df40a4b56bc1932a3e3552 : SUCCESS in 1h 07m 20s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/63193f7e86244c2fbb517e5c5f3f0b34 : SUCCESS in 46m 02s\n- ironic-inspector-tempest-discovery-2024.1 https://zuul.opendev.org/t/openstack/build/ebeb4b3f7d6d43bd93852e1e6dc5cc3e : SUCCESS in 28m 23s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/04a2ffa9d41f411aba34a90e36741c71 : SUCCESS in 28m 54s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/4e61f06aa5bb4ad8b08bccc5c4ddd8f2 : SUCCESS in 29m 43s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"01a862bb449c993262c5ea83e504a9b208e472ca","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-05-14 22:51:12.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"9d8e78fbcac39ad7f274512c4091c8cfbdf2e2f6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-05-15 15:14:12.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"6cd85ec056567c8ac000d883f7ea402cb408fb8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 15:15:29.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"6344b8afbb90bdc1aeee15f32dfe09248cbfb163","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 16:57:47.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a6fc3895a8d43f29141219088286b4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d8bc9c70b748a7b1eb9653129ffc40 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3df225f6c5245c7be94221ae617b233 : SUCCESS in 4m 20s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/4bdbe53073184dd5923a155a3a3f5fad : FAILURE in 1h 28m 53s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/02dd3e61998f4cf3bc314e973c39e61a : SUCCESS in 1h 05m 10s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/ac9ab1cfbc5849a08035a4400328499b : FAILURE in 1h 27m 08s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/423ef34677bf43889a65f61c98c0c223 : SUCCESS in 32m 24s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/cedd3559c3634f7cb14dccaa5325dcc6 : SUCCESS in 36m 56s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/3e6a56ee9c1442369c43c127b530dbbb : SUCCESS in 36m 51s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/0646af16170743848251d11743ffbfa0 : SUCCESS in 35m 07s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/93db481d8ddf4b7290da837fead22d6f : SUCCESS in 28m 49s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/51d804d22de04fe191fa4c68421bb3b9 : SUCCESS in 44m 06s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/297827a1b48540b28f620c0bf89a037e : SUCCESS in 1h 06m 45s\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/c3720e993c964c449a2162c14b9b2007 : SUCCESS in 1h 04m 57s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/56bc024a60a846f89956add6e83ae8ac : SUCCESS in 49m 33s\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/589c66ce41be45e599e62874f2409f3c : SUCCESS in 1h 05m 36s\n- ironic-inspector-tempest-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/51c8312b216b4249bc7a5f71765f23b7 : SUCCESS in 58m 54s\n- ironic-inspector-tempest-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/5605c9d368e94b379fc17d295269c90a : SUCCESS in 50m 27s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f15fe9fbc7a9409185b79891ea0737b9 : SUCCESS in 1h 33m 00s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/6a4e54602825452681a439d0bc874254 : SUCCESS in 1h 00m 41s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/7812a2869f5343bb8beb4b1da7373deb : SUCCESS in 57m 32s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/734562ff7d26483d93a31ff79a5931f3 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":17},{"id":"e0fdcf18d73558b59732b67f72704aaf603e8b2a","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-05-22 13:06:31.000000000","message":"Patch Set 17:\n\n(19 comments)","accounts_in_message":[],"_revision_number":17},{"id":"d2c7d2465f51f85628c6c70c46d9c07d6b4e1510","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-05-23 03:16:04.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"7849d027e0ffa8ba10435bcc96a177a9e85fde3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 05:11:52.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/984265023a8e4baf910ba6bdb80e83d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a52c6e9c21f4f45998b0212b4ffddba : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/608314fb67ad44a3bb9ad07a4df834ee : SUCCESS in 2m 09s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/35a0cb5923cb40d797af9c955e9443f7 : SUCCESS in 46m 34s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/11493743a6234c06a6fa2b8b9645535c : SUCCESS in 48m 45s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/3d83b3edbd5c4ea597830a46539a862d : SUCCESS in 1h 12m 37s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/72929de6faac46b39816b1a7d57d30ac : SUCCESS in 1h 09m 37s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/cc726a815aa8466282aa5f6285c8df30 : SUCCESS in 36m 24s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/608e4344975a4458b2bf1d92e7291de3 : SUCCESS in 22m 02s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/063f517d8370433eba266e7a5da9d363 : SUCCESS in 21m 29s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/e3ca1074cb2c4ff5bb6e8aae36541f2c : SUCCESS in 28m 01s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/5f3fa9e6eb8f4293afa15493c71c8fb3 : SUCCESS in 32m 11s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/1064487b9d04409bbc7865db82a75843 : SUCCESS in 32m 37s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b1b3c8503a8c4e8983b3f69c35d1282b : SUCCESS in 46m 54s\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/10cc790dcf4541bd8f4fbf4959f83737 : SUCCESS in 41m 04s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/ce8f9cfd8cd9415b9a2c13cf7d4be6fb : SUCCESS in 45m 36s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/e60a6fb797d24241a46f751fcb1d8c3f : SUCCESS in 55m 01s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/aeb3acb92d744254b484c24a9813919e : SUCCESS in 52m 53s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/0af5efbe2d1d49ea90af65c65b029f59 : SUCCESS in 43m 21s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/aae38508ef0d45c5a980bf7563a37dc5 : SUCCESS in 43m 53s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/5b6e12cf016044b999d414bc7028812a : SUCCESS in 46m 22s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/cd066eb87236407cbbd23a0cb75ff019 : SUCCESS in 1h 16m 07s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/096af8a4804d46f5add2034fbe4bda51 : SUCCESS in 54m 49s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/8108e7a1062f43a0b8f0543f566ac913 : SUCCESS in 1h 03m 55s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/3411c5fa4bdb4dc18336c368c0e0484f : SUCCESS in 1h 04m 23s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/042cc722f35d499eb1421ca20fb3de50 : SUCCESS in 1h 49m 57s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.1 https://zuul.opendev.org/t/openstack/build/948246aa3ab74b0296e31809e2e7b47e : SUCCESS in 1h 27m 14s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/5d551660c8f24bd9bb5f1d86d4be654c : SUCCESS in 1h 12m 10s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/6dc8d292ff154fee8f6e7bbac4af07be : SUCCESS in 1h 27m 52s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/16f8b641f8ee4aca905568acdeb0620d : SUCCESS in 43m 04s\n- ironic-inspector-tempest-discovery-2024.1 https://zuul.opendev.org/t/openstack/build/f3f9879b358546c8b72809bb1e5e7f30 : SUCCESS in 41m 12s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/064a9611cdbe49599662277702d8fc98 : SUCCESS in 36m 18s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/7b9c1c7c484b410ebd16e3f438897f84 : SUCCESS in 42m 59s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"6cef011a9f2388f535ad980b72521e1e7019a40f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 05:12:27.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"6753a2142c228e3c62e281b517c26bebf4990d09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 06:56:36.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4188bb05f9404c08b363d0cf0c8a323d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b9314acbac426a90ba18cc744792cc : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5a8bb8ce90844b2ba7db27aa7c5e3b7 : SUCCESS in 2m 48s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/7ef4d034a40343939d42520f3e3f0f64 : SUCCESS in 1h 10m 41s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/6b8a421d46634876988e43132b9251d3 : SUCCESS in 1h 00m 49s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/a0baaa3c6af1472fb68e9cab0a0beb37 : SUCCESS in 1h 03m 37s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/60a461312b874e0c8f0c19e8e9344e9e : SUCCESS in 38m 50s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/84e5b547010b43018e7f85f9f29ef442 : SUCCESS in 32m 23s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/135485afe7054c9da3e56a7e219adee6 : SUCCESS in 32m 17s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/618f4c3cab2242309ddc48248a1d2b92 : SUCCESS in 32m 04s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/d99baa67711644439b5d26f32b0647c0 : SUCCESS in 32m 50s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/a8c0bfc230e44f1c8f7219f81b899914 : SUCCESS in 26m 42s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/04ecf967f3df4214b53f4a4af5c9282b : SUCCESS in 46m 36s\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/f398458938234d17a9458fb3b7e0d96a : SUCCESS in 40m 02s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/0bb8ed305afb4e6bb3271ab7bec4a005 : SUCCESS in 50m 18s\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/cfccf020ece44666b33724a8e9e4b068 : SUCCESS in 48m 49s\n- ironic-inspector-tempest-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/31c55e98712d4e99b4567bd97d3ca4fd : SUCCESS in 51m 09s\n- ironic-inspector-tempest-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/3b3eaae46a05457ab603cd7c2376841f : SUCCESS in 41m 28s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/0dd4016871e3439d88b573114ec63ac5 : SUCCESS in 1h 38m 44s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/145ec6b84a514d0b8693a4a4ca2e3929 : SUCCESS in 42m 04s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/3d0027ea44c442abb1249b9f100eb4a3 : SUCCESS in 56m 34s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/2a176cacec074e54af0b591ce83d0c9c : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":17},{"id":"77db2c5d34a9e5670f5519b9e125618e71c97644","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 06:56:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"e6ff9ab5489029f10201bf4ed1bbfba2d3170f8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 06:57:31.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1860533d644428cbe208e9046ed2c9b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e455aead80d24255847b62e0efa4f6c9 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"5fdaffe11125cdb2a376c8cfb9fe973a78ed3caa","revisions":{"6a43dd20c8e1f9530f32fb68fd74862c0f0e3057":{"kind":"REWORK","_number":1,"created":"2024-03-21 19:13:45.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/1"}}},"commit":{"parents":[{"commit":"f5bfe9b319dc76500bd20c18037af01f67125312","subject":"Merge \"Test multiple boot interfaces as part of one CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/f5bfe9b319dc76500bd20c18037af01f67125312"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:13:42.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour of giving .json extensions a specially treatment.","message":"Patch to enforce json extension works in existing API behaviour\nof giving .json extensions a specially treatment.\n\nThis is me testing the tempest waters...\n\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/6a43dd20c8e1f9530f32fb68fd74862c0f0e3057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/6a43dd20c8e1f9530f32fb68fd74862c0f0e3057"}]},"branch":"refs/heads/master"},"201b7d9d800ba2bedeb76e071fd5050839feb7dd":{"kind":"REWORK","_number":2,"created":"2024-03-22 13:53:13.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/2"}}},"commit":{"parents":[{"commit":"f5bfe9b319dc76500bd20c18037af01f67125312","subject":"Merge \"Test multiple boot interfaces as part of one CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/f5bfe9b319dc76500bd20c18037af01f67125312"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-22 13:52:38.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour of giving .json extensions a specially treatment.","message":"Patch to enforce json extension works in existing API behaviour\nof giving .json extensions a specially treatment.\n\nThis is me testing the tempest waters...\n\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/201b7d9d800ba2bedeb76e071fd5050839feb7dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/201b7d9d800ba2bedeb76e071fd5050839feb7dd"}]},"branch":"refs/heads/master"},"159175b53cf598591585911707d095347b8a870d":{"kind":"REWORK","_number":3,"created":"2024-03-22 15:35:18.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/3"}}},"commit":{"parents":[{"commit":"f5bfe9b319dc76500bd20c18037af01f67125312","subject":"Merge \"Test multiple boot interfaces as part of one CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/f5bfe9b319dc76500bd20c18037af01f67125312"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-22 15:34:48.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/159175b53cf598591585911707d095347b8a870d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/159175b53cf598591585911707d095347b8a870d"}]},"branch":"refs/heads/master"},"af623d5d2f0a164223f01e4e1a1aa0fe458f7950":{"kind":"REWORK","_number":4,"created":"2024-03-25 20:23:49.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/4"}}},"commit":{"parents":[{"commit":"f5bfe9b319dc76500bd20c18037af01f67125312","subject":"Merge \"Test multiple boot interfaces as part of one CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/f5bfe9b319dc76500bd20c18037af01f67125312"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-25 20:23:30.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/af623d5d2f0a164223f01e4e1a1aa0fe458f7950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/af623d5d2f0a164223f01e4e1a1aa0fe458f7950"}]},"branch":"refs/heads/master"},"b8efe320172acee585e1997e5271229851beea1d":{"kind":"REWORK","_number":5,"created":"2024-03-26 13:45:14.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/5"}}},"commit":{"parents":[{"commit":"f5bfe9b319dc76500bd20c18037af01f67125312","subject":"Merge \"Test multiple boot interfaces as part of one CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/f5bfe9b319dc76500bd20c18037af01f67125312"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-26 13:44:43.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/b8efe320172acee585e1997e5271229851beea1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/b8efe320172acee585e1997e5271229851beea1d"}]},"branch":"refs/heads/master"},"0b7e80fbe3787f2bb61591fd8315319703f16b1c":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-03-27 17:14:17.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/26/913926/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/6"}}},"commit":{"parents":[{"commit":"f5bfe9b319dc76500bd20c18037af01f67125312","subject":"Merge \"Test multiple boot interfaces as part of one CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/f5bfe9b319dc76500bd20c18037af01f67125312"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-03-27 17:14:17.000000000","tz":0},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/913467\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/0b7e80fbe3787f2bb61591fd8315319703f16b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/0b7e80fbe3787f2bb61591fd8315319703f16b1c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9fea99ed49c7b93fe25eb0998b55d9c6e82daa72":{"kind":"REWORK","_number":7,"created":"2024-03-28 12:06:59.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/7"}}},"commit":{"parents":[{"commit":"f5bfe9b319dc76500bd20c18037af01f67125312","subject":"Merge \"Test multiple boot interfaces as part of one CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/f5bfe9b319dc76500bd20c18037af01f67125312"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-28 12:06:38.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/913467\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/9fea99ed49c7b93fe25eb0998b55d9c6e82daa72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/9fea99ed49c7b93fe25eb0998b55d9c6e82daa72"}]},"branch":"refs/heads/master"},"96117c9d97e6411403237cb0b1ca6bd004320881":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-03-29 16:56:30.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"real_uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/26/913926/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/8"}}},"commit":{"parents":[{"commit":"0eb9ae7445d6b6077d55c54ce9c4337e7d8ebc8a","subject":"Leverage the remote microversion without requiring human config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/0eb9ae7445d6b6077d55c54ce9c4337e7d8ebc8a"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-29 16:56:30.000000000","tz":0},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/913467\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/96117c9d97e6411403237cb0b1ca6bd004320881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/96117c9d97e6411403237cb0b1ca6bd004320881"}]},"branch":"refs/heads/master","description":"Rebase"},"05da8a8055b8dddeeca2228097558121c2cce8f5":{"kind":"REWORK","_number":9,"created":"2024-03-30 09:15:05.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/9"}}},"commit":{"parents":[{"commit":"0eb9ae7445d6b6077d55c54ce9c4337e7d8ebc8a","subject":"Leverage the remote microversion without requiring human config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/0eb9ae7445d6b6077d55c54ce9c4337e7d8ebc8a"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-30 09:13:30.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/913467\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/05da8a8055b8dddeeca2228097558121c2cce8f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/05da8a8055b8dddeeca2228097558121c2cce8f5"}]},"branch":"refs/heads/master"},"518a88d19af2e8d8d22506d33503f9c1aa843113":{"kind":"REWORK","_number":10,"created":"2024-04-05 14:13:26.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/10"}}},"commit":{"parents":[{"commit":"0eb9ae7445d6b6077d55c54ce9c4337e7d8ebc8a","subject":"Leverage the remote microversion without requiring human config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/0eb9ae7445d6b6077d55c54ce9c4337e7d8ebc8a"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-04-05 14:11:39.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/913467\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/518a88d19af2e8d8d22506d33503f9c1aa843113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/518a88d19af2e8d8d22506d33503f9c1aa843113"}]},"branch":"refs/heads/master"},"6c4f7015247b3b1f482fcddfae82f35949f9caaf":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-04-19 16:48:38.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/11"}}},"commit":{"parents":[{"commit":"c942d60b98b7104bd3aa502d999428f16c95d4ea","subject":"Merge \"Unprovision iso ramdisk boot from test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/c942d60b98b7104bd3aa502d999428f16c95d4ea"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-04-19 16:48:38.000000000","tz":0},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/913467\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/6c4f7015247b3b1f482fcddfae82f35949f9caaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/6c4f7015247b3b1f482fcddfae82f35949f9caaf"}]},"branch":"refs/heads/master","description":"Rebase"},"5a950b87f29b833ec49a5c7fe3c7f8e227b8e3c3":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-05-07 12:31:32.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/12"}}},"commit":{"parents":[{"commit":"926db2e70a09839ad4237a54a4e514c1acbf49b9","subject":"Merge \"Remove agent_ipmitool/pxe_ipmitool tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/926db2e70a09839ad4237a54a4e514c1acbf49b9"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-05-07 12:31:32.000000000","tz":0},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/913467\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/5a950b87f29b833ec49a5c7fe3c7f8e227b8e3c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/5a950b87f29b833ec49a5c7fe3c7f8e227b8e3c3"}]},"branch":"refs/heads/master","description":"Rebase"},"29a77c62ed7c345afb421484f70e6952f060a98b":{"kind":"REWORK","_number":13,"created":"2024-05-07 14:22:41.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/13"}}},"commit":{"parents":[{"commit":"926db2e70a09839ad4237a54a4e514c1acbf49b9","subject":"Merge \"Remove agent_ipmitool/pxe_ipmitool tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/926db2e70a09839ad4237a54a4e514c1acbf49b9"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-05-07 14:22:02.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/29a77c62ed7c345afb421484f70e6952f060a98b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/29a77c62ed7c345afb421484f70e6952f060a98b"}]},"branch":"refs/heads/master"},"9e96fcde8369a62a5124402156779c594fb0cba6":{"kind":"REWORK","_number":14,"created":"2024-05-07 15:49:16.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/14"}}},"commit":{"parents":[{"commit":"926db2e70a09839ad4237a54a4e514c1acbf49b9","subject":"Merge \"Remove agent_ipmitool/pxe_ipmitool tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/926db2e70a09839ad4237a54a4e514c1acbf49b9"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-05-07 15:44:46.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/9e96fcde8369a62a5124402156779c594fb0cba6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/9e96fcde8369a62a5124402156779c594fb0cba6"}]},"branch":"refs/heads/master"},"50747cab4aab425fd99e065fc1fb26dfc29784a2":{"kind":"REWORK","_number":15,"created":"2024-05-07 16:13:25.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/15"}}},"commit":{"parents":[{"commit":"926db2e70a09839ad4237a54a4e514c1acbf49b9","subject":"Merge \"Remove agent_ipmitool/pxe_ipmitool tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/926db2e70a09839ad4237a54a4e514c1acbf49b9"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-05-07 16:13:13.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extension is given a special treatment.\n\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/50747cab4aab425fd99e065fc1fb26dfc29784a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/50747cab4aab425fd99e065fc1fb26dfc29784a2"}]},"branch":"refs/heads/master"},"9b979190decd7ea5bcb1945d679d1e2b27a3ef04":{"kind":"REWORK","_number":16,"created":"2024-05-12 05:11:26.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/16"}}},"commit":{"parents":[{"commit":"926db2e70a09839ad4237a54a4e514c1acbf49b9","subject":"Merge \"Remove agent_ipmitool/pxe_ipmitool tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/926db2e70a09839ad4237a54a4e514c1acbf49b9"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-05-12 05:10:55.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extensions are stripped out during the search or while taking\nany action on a node through its name or uuid.\n\nThis pecan feature ensures that a reference to\n``0178-0c2c-9c26-ca69-3011-a9dd.json`` is treated\nas ``0178-0c2c-9c26-ca69-3011-a9dd``\n\nIf we had two nodes that differ by a .json extension in either of the names,\n``\"node\", \"node.json\"``, and we intend for the latter, we’ll need to append an\nadditional \".json\" extension in the reference query, else the node without the\nextension is treated as the default node, and the extension node is only\nconsidered if it’s the only node with the name.\n\nSame workaround if we wanted our ``0178-0c2c-9c26-ca69-3011-a9dd.json``\ncorrectly targeted, we append an additonal \".json\" extension to our query.\n\nThis patch tries to validate this behavior in API versions prior to 1.91 where\nthis feature exists.\n\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/9b979190decd7ea5bcb1945d679d1e2b27a3ef04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/9b979190decd7ea5bcb1945d679d1e2b27a3ef04"}]},"branch":"refs/heads/master"},"5fdaffe11125cdb2a376c8cfb9fe973a78ed3caa":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2024-05-12 05:19:48.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/26/913926/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/26/913926/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/26/913926/17"}}},"commit":{"parents":[{"commit":"926db2e70a09839ad4237a54a4e514c1acbf49b9","subject":"Merge \"Remove agent_ipmitool/pxe_ipmitool tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/926db2e70a09839ad4237a54a4e514c1acbf49b9"}]}],"author":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-03-21 19:08:54.000000000","tz":60},"committer":{"name":"cid","email":"afonnepaulc@gmail.com","date":"2024-05-12 05:19:34.000000000","tz":60},"subject":"Patch to enforce json extension works in existing API behaviour","message":"Patch to enforce json extension works in existing API behaviour\n\nCurrently, .json extensions are stripped out during the search or while\ntaking any action on a node through its name or uuid.\n\nThis pecan feature ensures that a reference to\n``0178-0c2c-9c26-ca69-3011-a9dd.json`` is treated\nas ``0178-0c2c-9c26-ca69-3011-a9dd``\n\nIf we had two nodes that differ by a .json extension in either of the\nnames, ``\"node\", \"node.json\"``, and we intend for the latter, we’ll\nneed to append an additional \".json\" extension in the reference query,\nelse the node without the extension is treated as the default node,\nand the extension node is only considered if it’s the only node with\nthe name.\n\nSame workaround if we wanted our ``0178-0c2c-9c26-ca69-3011-a9dd.json``\ncorrectly targeted, we append an additonal \".json\" extension to our\nquery.\n\nThis patch tries to validate this behavior in API versions prior to\n1.91, where this feature exists.\n\nChange-Id: I613e596e8cf52f36b9c7caac30bdb3ff0d3bf821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/5fdaffe11125cdb2a376c8cfb9fe973a78ed3caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/5fdaffe11125cdb2a376c8cfb9fe973a78ed3caa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
