)]}'
{"id":"openstack%2Fironic-tempest-plugin~927029","triplet_id":"openstack%2Fironic-tempest-plugin~master~I0f03575ed99afbfe49c8207403d4d0ae5d1d50da","project":"openstack/ironic-tempest-plugin","branch":"master","topic":"reenable-adoption","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2024-09-20 10:28:25.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I0f03575ed99afbfe49c8207403d4d0ae5d1d50da","subject":"The adoption test should not depend on the iscsi deploy","status":"MERGED","created":"2024-08-23 14:14:06.000000000","updated":"2024-09-20 10:29:28.000000000","submitted":"2024-09-20 10:28:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927029-reenable-adoption","meta_rev_id":"212d85de5e7fb7f888f798bc845660e6436e5153","_number":927029,"virtual_id_number":927029,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-20 10:28:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2024-09-19 15:48:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2024-09-19 14:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2024-09-19 15:48:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-23 16:13:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-19 14:38:46.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-09-19 15:48:38.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"c4482bb8b0afc712f1f93be7bc05279f0f4bfab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-08-23 14:14:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d171be7fbeac8375940075a7b78000941386f64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 16:13:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05679fd428674d878583246b86be9560\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd73d818bd84838b07d36f49eb2b0dc : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db01d18e3cf44dd2ac2acc2127d1a882 : SUCCESS in 4m 42s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/01084d65fbe74e269482adcc97887d03 : SUCCESS in 1h 15m 33s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/88e8b0d109e54024aa4201e2466404b1 : SUCCESS in 1h 28m 03s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/822da883fa8c467da3838d5cd0069fa2 : SUCCESS in 1h 27m 20s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/2213f348ac5e49a7a1b593e3ae57c466 : SUCCESS in 1h 26m 23s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b69cd2fbcefc437bb7a6bd6ef642c5d9 : SUCCESS in 33m 59s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/3ef3bd58a8f34025be830378c0c9a1ec : SUCCESS in 34m 01s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/cf4be3a9d5f340af8a5688ce19138dad : SUCCESS in 21m 52s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/522bb4bff7294f6e9eb0d088641d26d2 : SUCCESS in 55m 55s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/6edfd63df0384736bfa2d763e8dcb6f1 : SUCCESS in 30m 13s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/5028a1670d9c4fd78519290af0665a4a : SUCCESS in 49m 42s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/a371a5c957d2497ca65f84176a253879 : SUCCESS in 49m 17s\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/34dbb20f1f094b1b9d4e879c162aa9a5 : SUCCESS in 1h 18m 24s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/6a86c16e3aa945fab019b15619f413ab : SUCCESS in 1h 17m 02s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/ecf26934aba34d5c9a49db27079f715d : SUCCESS in 55m 07s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/000db1477def4790b0d98315adbbaeda : SUCCESS in 1h 07m 04s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/f158d989690b43ebb7b832e89f12df25 : SUCCESS in 1h 19m 31s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/f2b7aae7363545f08d7cdf26d98fa9da : SUCCESS in 46m 07s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/d08ee61b5a2f4a9fb1ca9021c584128f : SUCCESS in 38m 38s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/fef971ddc034435a94ea1a0b4e2a38fa : SUCCESS in 1h 36m 06s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/8173f41c603f4b23b4260a3324aea9c2 : SUCCESS in 1h 28m 49s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/75a31ed1ed0b400399b4dbfc3cb47a72 : SUCCESS in 1h 04m 49s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/0b626e31f86a448f82eec0b2d16b58d0 : SUCCESS in 1h 11m 09s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/b5b5b5a0c97147c6b6a46b832d10c827 : SUCCESS in 1h 46m 13s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.1 https://zuul.opendev.org/t/openstack/build/757966c45a324f358660bace5428e53f : SUCCESS in 1h 53m 56s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/b81191249e6f4921943f13dfb168f311 : SUCCESS in 1h 39m 52s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/e50832e90d56484da0a876654aa6e3f3 : SUCCESS in 1h 50m 54s (non-voting)\n- ironic-inspector-tempest-discovery-2024.1 https://zuul.opendev.org/t/openstack/build/4bb1e11dd86c462098ba6164c8dcc3d6 : FAILURE in 42m 49s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/c43a2c4342074538be9b5fec3903a542 : SUCCESS in 1h 05m 02s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/f41644e2e78648838fc4e1bfcc77ccaf : SUCCESS in 42m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28af4005d5ed2a5c934dc9166da7b1a45e8fcf17","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-08-29 15:39:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42ec9215adf137b7efa863afe17213cdd9a28e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 17:33:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6bc9e222d57486590f08089cdb963ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5599e7181e4f4e94a9cfe725a7b06199 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43993c70ba6148a681f6d08311c333dd : SUCCESS in 2m 39s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/c9452fb319874a068a9a52816875731a : SUCCESS in 1h 07m 12s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/c267939c23034892b4a769974ed5e22b : FAILURE in 1h 00m 43s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/14c38d17fff348dbbfe9cc32d4fde765 : SUCCESS in 58m 50s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/738f0c3de8414b608f3defc4322e621e : SUCCESS in 52m 51s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/45b6f1d4b6ca4bab85dc4402ec91c4af : SUCCESS in 24m 28s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/d8be1550930549f29aee4ab4ebd8960c : SUCCESS in 20m 14s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/c46b1cc23f47439fb709ec5d68091c20 : SUCCESS in 24m 38s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/fa96d3e2399341b1a079cd80b212bdd9 : SUCCESS in 22m 57s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/6dcf8195787b47ad927b9c4d1b1e930e : SUCCESS in 19m 41s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/828e14fe19994670ad9a6591d7324278 : SUCCESS in 19m 45s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/36ad22b16a6842fe801cfd07637f6e47 : SUCCESS in 48m 42s\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/7a0eed205152492aa1c2257fa99d8e2c : SUCCESS in 55m 16s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/7f09cf1dd5d04dadaabdfeefec7ff456 : SUCCESS in 51m 04s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/57dc7a0bf5f844349c14e015736a3b3c : SUCCESS in 52m 59s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/0ef3760a395a42a098a70fe48348f29c : SUCCESS in 38m 10s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/bb5d2e5ed15b430d9be1cb055ae7fdca : SUCCESS in 44m 48s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/691f5a3cd15c449594272b3cc747056c : SUCCESS in 46m 20s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/9f76fd3e46334abd8169dd6e1372e25d : SUCCESS in 44m 10s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/29f1e2b5a3354afc8e90d3a82c032810 : SUCCESS in 1h 47m 04s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/f72fe954b91e4a0cad564e1e47c965da : SUCCESS in 52m 14s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/560ba85874df4a3490eebeeb47e7c145 : SUCCESS in 46m 21s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/49287ea5de5641ae8024d866f54440d5 : SUCCESS in 1h 06m 34s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/356d2d0ee8e14e1db4db8457115de971 : SUCCESS in 1h 24m 08s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.1 https://zuul.opendev.org/t/openstack/build/38c8facb001140c5ab3de2d1f6e0aaf3 : FAILURE in 1h 32m 24s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/fefea4edf35e4bd7a0d2e05c9cb45672 : SUCCESS in 1h 25m 37s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/12ee9bca22184bf1891bb645066d57a1 : SUCCESS in 1h 18m 44s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/02e9357bdd554eeebd6a1e41e7ac253b : SUCCESS in 39m 33s\n- ironic-inspector-tempest-discovery-2024.1 https://zuul.opendev.org/t/openstack/build/b7a619efa3234b588f54b242ddc6fd57 : SUCCESS in 52m 34s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/1bafa9a8d7184f2db3d7f2dac84300e4 : SUCCESS in 40m 18s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/18b2008d7c2b405d8ed32f932db64812 : SUCCESS in 45m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6b3381fb43444084fd2d7af254035d49a481c7c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-09-04 11:58:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b8bb7f5c5465acf3103f8fcad9ecbde7ecb0089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 12:52:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cabb62cb0b14e318737f888a5e40e46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91fd3ad5b03f402a9f74052bed5da343 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd75bdb8e9e499cba66199cdc195e9c : SUCCESS in 2m 37s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/93c4b35f483d43ce9f2051c569682847 : FAILURE in 23m 23s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/107fc5faf8684153a238fbbea7e2e28e : FAILURE in 29m 13s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/94a4de3e957a4fc3b5f438626195fa35 : FAILURE in 21m 18s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/a437200f02c949bbaaf5fc327cde99ac : FAILURE in 28m 33s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b06c55603cd54404b1f989a7e6a7918b : FAILURE in 23m 49s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/65b3ba1c03ed4a5b819657c218bf485a : FAILURE in 17m 31s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/e4c0178147534041abb861ec30cf2572 : FAILURE in 24m 28s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/c8c2528ac52441f5b53f3cb1635ae4c9 : FAILURE in 16m 46s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/3a2ed141ee7d44958a192c3b53e9d755 : FAILURE in 26m 06s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/7a8b023fcff74341a252d0da1e6e2d94 : FAILURE in 24m 37s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/228fc50d21e448808d147e1d5ee3a98d : FAILURE in 25m 53s\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/4e986bf51608485fb785330cf7ac15a0 : FAILURE in 39m 42s\n- ironic-inspector-tempest-2023.2 https://zuul.opendev.org/t/openstack/build/2fee86cd68d642ab9cba2f2a29bf1960 : FAILURE in 29m 52s\n- ironic-inspector-tempest-2023.1 https://zuul.opendev.org/t/openstack/build/5a1e8bc46e2a49b697c62d3d0c0fadf0 : FAILURE in 41m 34s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/6461c7451ef246c086cf7adf6744a7e5 : FAILURE in 30m 57s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/3fa4ccbce4314fb48137e0bf3e050c0e : FAILURE in 22m 31s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/6b67dd1797ee4be5a2f7f111c268fd44 : FAILURE in 33m 33s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/98cd945cef3d43c9b1420ea68c7eaeed : FAILURE in 31m 43s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/ad1ef4de047a4498a9a1cdb3217e7f9f : FAILURE in 32m 37s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/71b64fc2044a4c0ebbc38cc7a7239d24 : FAILURE in 32m 31s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/72f270d439c945be8481651e853381df : FAILURE in 30m 06s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/f4c9254526ac4253a7861e27313d22b5 : FAILURE in 27m 31s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/40a41d96a98d42c5ac9d262b2735fd08 : FAILURE in 47m 09s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.1 https://zuul.opendev.org/t/openstack/build/04cf302e1a6a41b5aebff0347205bcd3 : FAILURE in 45m 10s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/1a581467e72a41e7a3e580069f8828f4 : FAILURE in 41m 31s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/09d52ccb28294f79ab65b68936ac50e5 : FAILURE in 48m 43s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/a71542f1e4a6442c896d296f97a66b44 : FAILURE in 27m 59s\n- ironic-inspector-tempest-discovery-2024.1 https://zuul.opendev.org/t/openstack/build/824c98e38e6049eea0ef8fe0f7ae8019 : FAILURE in 30m 30s\n- ironic-inspector-tempest-discovery-2023.2 https://zuul.opendev.org/t/openstack/build/fba4cce34d394e8d92edc6aa8e524884 : FAILURE in 39m 11s\n- ironic-inspector-tempest-discovery-2023.1 https://zuul.opendev.org/t/openstack/build/34ef670add7a42d5a890c1d00f59f069 : FAILURE in 42m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5797856c635218e18273fda79cd10d0fc3736df6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-09-19 11:36:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"29fa04b2f99763e65304bb2c5b300835a98d4e29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 13:42:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69a2cacfd68d4ad29711bc253c9b6879\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b819afff4521400e8b03b8a36305fb30 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a28fb8c60b514ddb825dd7d2b1780d2c : SUCCESS in 2m 10s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/59d658d19c6341e98721b6144b71f3a0 : SUCCESS in 55m 52s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/5f490996db154b618d5c329902b075ff : SUCCESS in 1h 14m 10s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/fa6eaa0bfc0646fabfab49f784abe201 : SUCCESS in 58m 27s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/9677b8b054694091869af4f3b0305367 : SUCCESS in 46m 40s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/1d302546542447d4b93896096293256d : SUCCESS in 23m 38s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/7a533ff77d8b4ee5b60918a510927a06 : SUCCESS in 20m 54s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/c09f9b144d6449c480d7ce80e3d3fa2c : SUCCESS in 21m 03s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/462bd86e727f454fb1db4377747ae461 : SUCCESS in 23m 18s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/6d6479df7d65495e9d2a67554f42e3d0 : SUCCESS in 21m 53s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/c3842018ec2c4f309c783447b61c82fb : SUCCESS in 31m 49s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/daab35fae0ff4c48bdd69bb5ddc74a4a : SUCCESS in 34m 48s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/a69a25ddf4834282a5238b6285a9532f : SUCCESS in 45m 52s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/4bbae72c54654e4282d1421ededf52b3 : SUCCESS in 49m 26s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/647d151c1f4a46c4b7e6085a6fd766e7 : SUCCESS in 54m 18s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/f24ad6952e7242909e1b5c7703144948 : SUCCESS in 52m 00s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/6285566839a34f0ab386739241d320fd : SUCCESS in 1h 08m 49s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/af18a3fffe724504b0b5bd95d4ebe24f : SUCCESS in 59m 29s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/90515e15ea5e4fcbadb5f0ca6f81932e : SUCCESS in 54m 55s\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/d1ff6c87d1fa4d1783cb15fd3eb5a03f : SUCCESS in 55m 57s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/5cedabb8abc14537bbf8f1a23b42b07e : SUCCESS in 1h 08m 10s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.1 https://zuul.opendev.org/t/openstack/build/a653899d850d4183980197b8fb4320ef : SUCCESS in 53m 53s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/0989604d95034a9c900fff8564995780 : SUCCESS in 1h 58m 31s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/ecfe41c892f24ab5ba25a269bde83e23 : SUCCESS in 1h 16m 56s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/c540dd2733d44653860ee569a3a35100 : SUCCESS in 47m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"913899e282d42b0adf2d1a5ba30a1489dfb58d60","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-09-19 14:38:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bb61a59f20f7abc85e856ecdae7f4edd83aafcd5","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-09-19 15:20:45.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"b0f618a06c41e358b89672bee86a2b3f05d22921","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-09-19 15:48:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1efbfd3636c4242f2a1e4a176c256057767618d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 15:49:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"99e6f005bcd65de5b210f1894bf7f750c75b52d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 17:25:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02e3c25e2da1419e9c4ebbd4c55c3a5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241338e2790349058560cbd0f7ace058 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a9a7f3b1604e0a91caabde3b34e5d5 : SUCCESS in 2m 28s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/e612df6afb514719bc638813de73c235 : SUCCESS in 1h 01m 53s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/936e987dd7ab43f5838cf70643afa2e3 : SUCCESS in 1h 01m 22s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/6a93c3a94daf43f6a217cb2150f5cca4 : FAILURE in 20m 17s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/34585d506f644a61a7e5c7250ed83f9c : SUCCESS in 25m 45s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/1027929fb17b45218d344a0311f73535 : SUCCESS in 23m 35s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/1e1ce666971e47df9bc862c8516f9152 : SUCCESS in 29m 51s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/6e0dcde3cffd4c27a472c3fabb2dc27b : SUCCESS in 22m 40s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/0fa87ed7d8494fb3b7920c8555213d3e : SUCCESS in 31m 39s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/4510df283f7d46ec813ba0853b907bf2 : SUCCESS in 35m 01s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/9ec38fccdcc845828b7f5811842b1b1e : SUCCESS in 42m 17s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/57ec6250fb984b979b90d33dcc8f0729 : SUCCESS in 37m 05s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/dcd3762ed7414df6af533252f9944053 : SUCCESS in 1h 31m 50s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/aad7b93de5d340da8e564c607de7e5fb : SUCCESS in 43m 50s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/894483d4fbca46cd808fff35c6778eb3 : SUCCESS in 1h 01m 25s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/f1ee701af44d4a7aa97344e4acc0b0e3 : SUCCESS in 44m 19s","accounts_in_message":[],"_revision_number":2},{"id":"aa6f86d603a0b7ed31d847b66032953a61f46857","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-09-20 07:17:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"119836584282cceefd2b69e83cb6bb4bf1459637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 09:02:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/945a8df281ae43499731c49bfce264b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cebec5bd6b624980ae1c3ded415bf33f : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d91423dc9f864648adf7393ee3c7a48f : SUCCESS in 2m 26s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/2dd6aedd4fc648678804917a36be8f54 : SUCCESS in 49m 58s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/73306634ad4a46babab04b2f198024b6 : SUCCESS in 1h 02m 18s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/0f5c9769e01347f0a203fab146a746e3 : SUCCESS in 44m 11s\n- ironic-standalone-2023.1 https://zuul.opendev.org/t/openstack/build/ac77fd646b554014951caadc6c7a803b : SUCCESS in 59m 34s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2a08f2fc9e3e46ef847b7b645ae20b4d : SUCCESS in 35m 31s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/3a6da6bf84574cc4836f68276030344a : SUCCESS in 23m 04s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/03ebfdce044a448d995a00b059fad6f0 : SUCCESS in 21m 46s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/55e8fc2ecb5c448ea35aae5555c4b251 : SUCCESS in 27m 13s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/59c4366092ee4abda306a8c41c12cf6e : SUCCESS in 27m 52s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/e6509fad126646b9ad7eebef3a52953a : SUCCESS in 31m 10s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/d6734a8832434bb18b05dc83297de767 : SUCCESS in 44m 46s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/11f42fef9caf4d56b64b92750b317e8f : SUCCESS in 43m 32s\n- ironic-standalone-anaconda-2023.1 https://zuul.opendev.org/t/openstack/build/569a69727ed04029a39411cfd2e19a4c : SUCCESS in 44m 41s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/062ada091a24476998cd71a4791b189d : SUCCESS in 1h 39m 13s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/4814910fba5246a8a8202e04a429db2b : SUCCESS in 1h 04m 39s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/44448bc91be14701953a7a8134138d26 : SUCCESS in 54m 00s\n- ironic-standalone-redfish-2023.1 https://zuul.opendev.org/t/openstack/build/ee698de19bee4cbfaefd4af412616db9 : SUCCESS in 1h 02m 38s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/d5363a90a12a4996b660891ed4541409 : SUCCESS in 51m 39s\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/fbbaa7cb719246af9acff201a7d76f1c : SUCCESS in 43m 44s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/839352553c8b4361940f43694e1292e4 : SUCCESS in 1h 32m 11s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.1 https://zuul.opendev.org/t/openstack/build/b1425dbe8cbf449b8c9e1df6c447a86a : SUCCESS in 1h 26m 32s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/0660231b00104b26bb0b051ff93b40e1 : SUCCESS in 1h 12m 10s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.1 https://zuul.opendev.org/t/openstack/build/0420d105fbcf4317af871d53d881e5a3 : FAILURE in 1h 17m 59s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/2ce6d86d19574fe5b6ce2bf7b71e8920 : SUCCESS in 30m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e5fa4f8b5452c0e83161847b9ebc33270a44b0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 09:03:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"beb1a9ecd0288828807719006f240092d21458cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 10:28:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/489db6d8395b4c0abf46d4d02a90b82f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c938463bb95a4812adbe4f796d8bdb60 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a144f555e9437786ecd80ca8a17bd7 : SUCCESS in 2m 54s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/229f3ecaff7d406fa53ae8321fa9ab85 : SUCCESS in 39m 22s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/bf744924a1294afb8ae7064fc1c7289c : SUCCESS in 1h 02m 13s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/0dfbd5b8914a40df86bb97ed9a480044 : SUCCESS in 1h 02m 35s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/39f1760e5532483cb366ce0002e72e7a : SUCCESS in 36m 02s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/2935da9767a54125b4f33fb3c42a8ee6 : SUCCESS in 32m 40s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/b9f514dedb3e4b62bc3e7c3f9dbe86f6 : SUCCESS in 30m 46s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/cae24e889cd443a6b235f7f6691dfe28 : SUCCESS in 36m 26s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/4f8a81d9d8ee4f9a91febb678909f69f : SUCCESS in 31m 03s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/bd80a674408f4d2a9bdfb464e924461e : SUCCESS in 41m 20s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/9473d5ce9f004792b742421aaedcb3ee : SUCCESS in 48m 06s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/40ce12627999406aa7894acd77d80073 : SUCCESS in 48m 32s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d470db0dc2ab4afa8e9a4be37d81f0ed : SUCCESS in 1h 18m 56s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/f14478157f6642ab8286a1a9c0a8b0bb : SUCCESS in 56m 01s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/ff1e1c697f814e51b713386bd2d70907 : SUCCESS in 46m 57s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/67c299f01df144ca971edd76d5a8b95e : SUCCESS in 48m 41s","accounts_in_message":[],"_revision_number":2},{"id":"6f7ee50d3b300ac354c68ae888ed8ae857475cd7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 10:28:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"212d85de5e7fb7f888f798bc845660e6436e5153","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 10:29:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fa88d351a304c009c536f32276bd6f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dda7b0d57c9545309831a51d73d8a0ea : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f51e00f4aa190fac9e20a96e5bdf762d549497ba","revisions":{"fb0f8657c281d2b4a7f4d8b3c1e1ae914f17758d":{"kind":"REWORK","_number":1,"created":"2024-08-23 14:14:06.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/29/927029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/29/927029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/29/927029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/29/927029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/29/927029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/29/927029/1"}}},"commit":{"parents":[{"commit":"6d1e5510bff12c62e15e0b3e9ec3a7dc87ea752d","subject":"Merge \"Handle invalid configuration for vif test execution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/6d1e5510bff12c62e15e0b3e9ec3a7dc87ea752d"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-08-23 14:13:27.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-08-23 14:13:27.000000000","tz":120},"subject":"The adoption test should not depend on the iscsi deploy","message":"The adoption test should not depend on the iscsi deploy\n\nIt was removed many releases ago, the test is now always skipped.\n\nChange-Id: I0f03575ed99afbfe49c8207403d4d0ae5d1d50da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/fb0f8657c281d2b4a7f4d8b3c1e1ae914f17758d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/fb0f8657c281d2b4a7f4d8b3c1e1ae914f17758d"}]},"branch":"refs/heads/master"},"f51e00f4aa190fac9e20a96e5bdf762d549497ba":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-19 11:36:35.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/29/927029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/29/927029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/29/927029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/29/927029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/29/927029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/29/927029/2"}}},"commit":{"parents":[{"commit":"c9fc1882bc212968d47a216fc29310c74d43bd45","subject":"Merge \"CI: scale back the number of inspector and scope enforced jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/c9fc1882bc212968d47a216fc29310c74d43bd45"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-08-23 14:13:27.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-09-19 11:36:28.000000000","tz":120},"subject":"The adoption test should not depend on the iscsi deploy","message":"The adoption test should not depend on the iscsi deploy\n\nIt was removed many releases ago, the test is now always skipped.\n\nChange-Id: I0f03575ed99afbfe49c8207403d4d0ae5d1d50da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/f51e00f4aa190fac9e20a96e5bdf762d549497ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/f51e00f4aa190fac9e20a96e5bdf762d549497ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
