)]}'
{"id":"openstack%2Fironic-tempest-plugin~941054","triplet_id":"openstack%2Fironic-tempest-plugin~master~I415aff416a07a7b69baf8b152a37c3602d9702bc","project":"openstack/ironic-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2025-02-11 23:33:40.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-02-11 23:33:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I415aff416a07a7b69baf8b152a37c3602d9702bc","subject":"Improve check_vm_connectivity","status":"MERGED","created":"2025-02-08 18:03:33.000000000","updated":"2025-02-11 23:34:41.000000000","submitted":"2025-02-11 23:33:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941054","meta_rev_id":"39e4ab26b3eb6faee9fd37c8f4790193b974000d","_number":941054,"virtual_id_number":941054,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 23:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2025-02-11 16:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2025-02-11 22:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-02-11 22:30:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-08 22:05:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-10 14:12:43.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2025-02-10 16:23:47.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2025-02-11 16:36:46.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-02-11 22:30:50.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"534462414ad638d36d6792f0a32abe37f7256ac6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-08 18:03:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0138e9ac7746d18f09be1e59fbd794c78569eb2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-08 19:00:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3fa805fe69fc2de137c480f95e0db10179475d1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 22:05:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f204f5b989e44c48887fb3899ba260d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277734670a9e432883621d7f190c6ab5 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b74cdc210df44d6b41d0a370becdb99 : SUCCESS in 2m 09s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/7aa5e264d286497bbe7ffbd0435d769e : SUCCESS in 36m 51s\n- ironic-standalone-2024.2 https://zuul.opendev.org/t/openstack/build/5a8b1ab29c8a434b98767a24c24794e2 : SUCCESS in 46m 36s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/e61b63d8cc1643e78ad18d9cc1254245 : SUCCESS in 41m 33s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/ba1f56869b5f45ccb7bd9e0da8910435 : SUCCESS in 46m 44s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0cce7a9bc19e47b6aec431f83ccf6b25 : SUCCESS in 26m 00s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/6c12f74c13964d81ba3de44bf0a4986c : SUCCESS in 24m 59s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/2773f9325da9483bac690cf5dc9ba902 : SUCCESS in 20m 39s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/5121f52d161f4ed3862df8716b9c2250 : SUCCESS in 32m 13s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced-2024.2 https://zuul.opendev.org/t/openstack/build/21a7b742b1d04c4b85e44ba95f59a540 : SUCCESS in 35m 11s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/e4eedaa49c6c4747800ef997c9a7e100 : SUCCESS in 35m 05s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/972d7c21449e4050bdc60bc39c91e064 : SUCCESS in 20m 37s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/e1fe7bb777b048a68e9d60848b5d7f03 : FAILURE in 1h 42m 11s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/4ca67af9faa347729831fef789e7d1fd : SUCCESS in 47m 20s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/f0596e13932445039dfbf60e3681a4f3 : SUCCESS in 32m 15s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/0c7df8396d3345d6b2e7ac75daf3e224 : SUCCESS in 51m 17s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/cc3dccb07d8b44c390a32d179842ce21 : SUCCESS in 53m 14s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/0beb95568aaa46b58d5b8511795815ad : SUCCESS in 1h 09m 33s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/6121a3d2689642f0ae8207a6582df4a7 : SUCCESS in 39m 12s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/aaaa74335ad549d3b2e626ea5c969658 : SUCCESS in 1h 23m 10s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/f6ef807b8f8a438a8563973e31b38bcc : SUCCESS in 41m 57s\n- ironic-inspector-tempest-2024.2 https://zuul.opendev.org/t/openstack/build/2b22d74621264fcda17a9150c79a5c60 : SUCCESS in 28m 35s (non-voting)\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/24630737be9f4304b37e3b9eae083d67 : SUCCESS in 55m 11s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/dc0ee8e308ad448083dcc081f97da39e : SUCCESS in 1h 30m 22s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.2 https://zuul.opendev.org/t/openstack/build/fcb9b5ca99644da18b658dc6f5ec1839 : SUCCESS in 1h 06m 45s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.1 https://zuul.opendev.org/t/openstack/build/4875ad86f8f94f89b75b1ce081e6ec79 : SUCCESS in 1h 47m 36s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/5983e1c5bbea4f6eac8651112c43f3b1 : TIMED_OUT in 3h 03m 09s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/298469070aa74e73955782c889d2f420 : SUCCESS in 43m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"776334a66ccf8f6b7da4f1643fd815f2f9be3da8","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-09 07:27:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3aac2f4ce9a8e444a8fe6a90b837dd03af6c4e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-09 09:25:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75375cfd3f2f4d72b1804003f3d930e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1703051178774d3e83d0b89dd010cae9 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe140b07a95d47f09ed1bc6759e39d6a : SUCCESS in 2m 32s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/5733278902e24f0c81e2dff0ed323d92 : SUCCESS in 1h 02m 13s\n- ironic-standalone-2024.2 https://zuul.opendev.org/t/openstack/build/c338ec40ab674e50a4df3b6c954ebdbe : SUCCESS in 1h 01m 31s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/980454b9ed1f4f0cb21f35dd9d868582 : SUCCESS in 50m 53s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/c54736ea0c744077af1d8422f218c0a2 : SUCCESS in 38m 57s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/836ee70393464f5f891f78939179d0ce : SUCCESS in 22m 54s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/1d969bbc0a2441c4ba3838ec69c132c5 : SUCCESS in 22m 30s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/4459606e850048459d3e35cbba2ac96c : SUCCESS in 18m 30s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/360942cacd814a5cbbbdf69352419ce2 : SUCCESS in 18m 36s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced-2024.2 https://zuul.opendev.org/t/openstack/build/b15e8abd6eb44da38e2ef7195a8716f3 : SUCCESS in 21m 21s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/f48ef32e82ff4f95b6bb2e35182db8ed : SUCCESS in 19m 05s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/ff2b1defb77c46b8aa828bb80833de87 : SUCCESS in 18m 40s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/ad11b78e0b66434ea0dbce810040d20a : SUCCESS in 28m 43s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/27289b9fd3ca486792621e2d791791c3 : SUCCESS in 27m 37s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/af55374cbc7f4b7284a2f20c8e6bf104 : SUCCESS in 32m 37s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/bd15b059382d40ddbd128c2ac43ce9aa : SUCCESS in 29m 37s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/4ef2caef885e418c8c2405008574885a : SUCCESS in 1h 25m 16s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/84088797903f4c15b278fe360b3228b4 : SUCCESS in 1h 09m 24s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/9afa5b9c3c2f42948b05c393891ba228 : SUCCESS in 1h 15m 20s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/3a2590b4ca2c471493b2a569506505ce : SUCCESS in 1h 06m 14s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/db15b483eaf24e3fb69d9c76a9f8223c : SUCCESS in 59m 50s\n- ironic-inspector-tempest-2024.2 https://zuul.opendev.org/t/openstack/build/f054b1a94d8d4d06b24ace25bf8a6654 : SUCCESS in 53m 25s (non-voting)\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/23995da3f0754bae8fe140a166f22e23 : SUCCESS in 55m 14s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/814e2e504a6649b89050ba2ba9447ecd : SUCCESS in 1h 47m 44s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.2 https://zuul.opendev.org/t/openstack/build/4057841c56be4c6595c81d1fe592beb0 : SUCCESS in 1h 32m 52s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.1 https://zuul.opendev.org/t/openstack/build/81b8652b3a774ab291dad4f68a21c265 : FAILURE in 1h 38m 19s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/d8f50e2d55c74cf2b99bf104ea0e4e4c : SUCCESS in 1h 49m 04s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/ebedd70fa4c448598f60ffbe334e7b15 : SUCCESS in 45m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1231e1698084ca2a1071aa5332f8532a5874ef12","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-02-10 14:12:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"907986b6a20211eb764230ac9053034b5960108a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-10 15:19:00.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a7a494fcc37bb981a61e7e7758c6fbf668b34d91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-10 15:21:27.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"f06db6fd8880b8a1e0a674b6bb300d5a189457c9","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-10 15:21:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"16bb0c269e970567b85acd3648de830980fe7717","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-10 15:28:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"61c977faee86c4577c9b0fc9eca4089b147c2b8b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-02-10 15:41:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9865f1af887b88dd93e7e86b64014b7997c5466c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-10 15:43:47.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"2936f2d306cd5b055c78127d9af438555ed990f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-10 16:13:00.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"cfc04a1ca4998be3ab893c4767772ff7d27c9b23","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-10 16:13:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"43aef907f96dfd89e517ceb0350c4f31104c0071","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-02-10 16:23:47.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"278cdb3e2e2b72ca299957a76bf826213c5d43dc","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-10 17:14:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ee28fee88ab1bd83cbb347c95ad42512495d74be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-02-10 17:25:04.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"af347c213f17a8963a1714c2930e50cd3a969116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 19:40:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d178cd282594e6d82752cd9b7bd856b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d927a3e2e7c4a3d9ca033be0ae172c1 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b9a6ffa423f448b8f961c85190d14fc : SUCCESS in 5m 16s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/fa5a256b26764653aae98b92857b5458 : SUCCESS in 35m 29s\n- ironic-standalone-2024.2 https://zuul.opendev.org/t/openstack/build/d6d5f1d133ad4037aae4359b29737b34 : SUCCESS in 49m 34s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/4815620ba7e34500beef1b0ad2994614 : SUCCESS in 41m 32s\n- ironic-standalone-2023.2 https://zuul.opendev.org/t/openstack/build/dfd5a79fe29a4ee5919ed10bf51cb169 : SUCCESS in 32m 53s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/24652352f1744d70816e1df60fce070d : SUCCESS in 28m 23s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/cb7d6e2c7ace42799bf4492b20a6dd0e : SUCCESS in 22m 25s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/73358253d446452aa815724e5cfe6540 : SUCCESS in 27m 23s\n- ironic-tempest-functional-python3-2023.2 https://zuul.opendev.org/t/openstack/build/611041a125d54af39e88e42f7ffae10e : SUCCESS in 17m 33s (non-voting)\n- ironic-tempest-functional-rbac-scope-enforced-2024.2 https://zuul.opendev.org/t/openstack/build/590021ff68b74e299fcae03c3f9270b7 : SUCCESS in 36m 08s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/e37faf3650f34b44b9642f69dd49645f : SUCCESS in 38m 58s\n- ironic-tempest-functional-rbac-scope-enforced-2023.2 https://zuul.opendev.org/t/openstack/build/39dba6b792a5492985caee48b9d9686a : SUCCESS in 30m 09s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/d16cddb64224442090ee0d1f7c8451e5 : SUCCESS in 34m 28s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/37fbccbe864744d2b15b5fce7264c4e6 : SUCCESS in 39m 35s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/d1f99961d92d4d8fa65595b152d0650b : SUCCESS in 55m 50s\n- ironic-standalone-anaconda-2023.2 https://zuul.opendev.org/t/openstack/build/43711c444d54418f8b08c31bbe778e1f : SUCCESS in 51m 06s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/7258ebf51f1040058699c9afac02a831 : SUCCESS in 1h 32m 51s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/452684d3c0624f8c8ffef33b717af017 : SUCCESS in 49m 57s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/f1a86da6f9db4e3d88f7aefe05383b52 : SUCCESS in 56m 47s\n- ironic-standalone-redfish-2023.2 https://zuul.opendev.org/t/openstack/build/43661cad54b0407e9cd36c5e61e69b50 : SUCCESS in 38m 36s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/c4fc8ef30b6c40feadb2bcc796398294 : SUCCESS in 45m 26s\n- ironic-inspector-tempest-2024.2 https://zuul.opendev.org/t/openstack/build/0dc5dcab944e481ba1528e2a8665eb76 : SUCCESS in 42m 22s (non-voting)\n- ironic-inspector-tempest-2024.1 https://zuul.opendev.org/t/openstack/build/8519ad34e4d24c96a2eb4348e46e8a26 : SUCCESS in 46m 58s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/37c224e5618e466b8f5c32985759663b : SUCCESS in 1h 28m 28s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.2 https://zuul.opendev.org/t/openstack/build/65876b5dc2df4677b74fd7be75cf9c91 : SUCCESS in 1h 30m 53s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2024.1 https://zuul.opendev.org/t/openstack/build/52a5b7d3d92d42a2853c9a9ab7a5217a : SUCCESS in 1h 30m 05s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2023.2 https://zuul.opendev.org/t/openstack/build/06594650bfe54ebc800ef54e622ff03a : SUCCESS in 2h 09m 09s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/32b7d6c0b9464534a90d944500a9c162 : SUCCESS in 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bda57b2f7b4aec863f364e9af5793b54a80a47ac","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-02-11 16:36:46.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"826888ce6a029dbac8bbc52c62f8b0a32b7abdeb","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2025-02-11 22:30:50.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e0484878ce9e5ee4eba128d652daa71bc5fa090f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 22:31:23.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"6bd18a3a475f764439dcbfcb0721ff3d10445bc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 23:33:40.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f19ddc0d0f5f4e5080fc489bc8b893de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f32218b465e488fa61508660d93c120 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2cd47462bb547399e0e8d59554f6caf : SUCCESS in 3m 56s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/3397306ae5324957a7cf62c389a32285 : SUCCESS in 39m 34s\n- ironic-standalone-2024.2 https://zuul.opendev.org/t/openstack/build/f85a5af73a4246bcb25d02963710ff59 : SUCCESS in 46m 37s\n- ironic-standalone-2024.1 https://zuul.opendev.org/t/openstack/build/ad7a6df0fd5743ac834fb8a4c8faefb5 : SUCCESS in 50m 07s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2213e2228f054b0492fbdae3d3278a22 : SUCCESS in 38m 49s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/947bc7945fc2428b9aefa382d37b9160 : SUCCESS in 18m 28s\n- ironic-tempest-functional-python3-2024.1 https://zuul.opendev.org/t/openstack/build/afd0661fde26449bbf0aec10a3fc1657 : SUCCESS in 18m 56s\n- ironic-tempest-functional-rbac-scope-enforced-2024.2 https://zuul.opendev.org/t/openstack/build/7019c37295ff4cef9b143dac049e4f8d : SUCCESS in 25m 30s\n- ironic-tempest-functional-rbac-scope-enforced-2024.1 https://zuul.opendev.org/t/openstack/build/dd1483b6fdc442a58d0ce735e97828bb : SUCCESS in 17m 09s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/e4dec648616a436a898042ce574966b7 : SUCCESS in 48m 37s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/efd48cdca6d5485d934eec5a2c6090ad : SUCCESS in 33m 14s\n- ironic-standalone-anaconda-2024.1 https://zuul.opendev.org/t/openstack/build/05e1937336be4d168321b32c3d24bff5 : SUCCESS in 32m 27s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f3bb909400b24157907757dbb64dca61 : SUCCESS in 1h 01m 47s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/2055af47349d40b69d70d4cbd7c75e54 : SUCCESS in 30m 59s\n- ironic-standalone-redfish-2024.1 https://zuul.opendev.org/t/openstack/build/77a3381674cd4ab5b1df236de5528c30 : SUCCESS in 51m 29s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b1c1f309e74d4155a140978fa5bffdfd : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":7},{"id":"d3720358412352b58bd1b5b87033c4bc4785db28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 23:33:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"39e4ab26b3eb6faee9fd37c8f4790193b974000d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 23:34:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7e3f820058d43528214f14046f2b107\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2938a121e6444be865a37be97c14a84 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"514d3947d35a5d86f40ed3997ec2a680d2bbf9bd","revisions":{"0243884cb60d7e12913f60e58cc4c45b3fc6a39c":{"kind":"REWORK","_number":1,"created":"2025-02-08 18:03:33.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/54/941054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/54/941054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/1"}}},"commit":{"parents":[{"commit":"26bf3b501443ad9592446aac702afb4e6ca0b8b6","subject":"Remove reboot from basic baremetal ops","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/26bf3b501443ad9592446aac702afb4e6ca0b8b6"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-08 18:01:59.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-08 18:02:22.000000000","tz":0},"subject":"Improve check_vm_connectivity","message":"Improve check_vm_connectivity\n\nIt may take some time for cloud init to setup ssh keys,\nso its possible server start responding over ping, but SSH\nis still not awailable. This patch adds a check that we can\npass SSH authentication before doing final vm connectivity\ncheck.\n\nChange-Id: I415aff416a07a7b69baf8b152a37c3602d9702bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/0243884cb60d7e12913f60e58cc4c45b3fc6a39c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/0243884cb60d7e12913f60e58cc4c45b3fc6a39c"}]},"branch":"refs/heads/master"},"d4e1d3e052a1efe6eb8777298a6303516e2e071a":{"kind":"REWORK","_number":2,"created":"2025-02-08 19:00:12.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/54/941054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/54/941054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/2"}}},"commit":{"parents":[{"commit":"26bf3b501443ad9592446aac702afb4e6ca0b8b6","subject":"Remove reboot from basic baremetal ops","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/26bf3b501443ad9592446aac702afb4e6ca0b8b6"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-08 18:01:59.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-08 19:00:11.000000000","tz":0},"subject":"Improve check_vm_connectivity","message":"Improve check_vm_connectivity\n\nIt may take some time for cloud init to setup ssh keys,\nso its possible server start responding over ping, but SSH\nis still not awailable. This patch adds a check that we can\npass SSH authentication before doing final vm connectivity\ncheck.\n\nChange-Id: I415aff416a07a7b69baf8b152a37c3602d9702bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/d4e1d3e052a1efe6eb8777298a6303516e2e071a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/d4e1d3e052a1efe6eb8777298a6303516e2e071a"}]},"branch":"refs/heads/master"},"175b3adeda8668f82a4c31ed1ac790cfff32e606":{"kind":"REWORK","_number":3,"created":"2025-02-10 15:19:00.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/54/941054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/54/941054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/3"}}},"commit":{"parents":[{"commit":"26bf3b501443ad9592446aac702afb4e6ca0b8b6","subject":"Remove reboot from basic baremetal ops","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/26bf3b501443ad9592446aac702afb4e6ca0b8b6"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-08 18:01:59.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-10 15:19:00.000000000","tz":0},"subject":"Improve check_vm_connectivity","message":"Improve check_vm_connectivity\n\nIt may take some time for cloud init to setup ssh keys,\nso its possible server start responding over ping, but SSH\nis still not awailable. This patch adds a check that we can\npass SSH authentication before doing final vm connectivity\ncheck.\n\nChange-Id: I415aff416a07a7b69baf8b152a37c3602d9702bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/175b3adeda8668f82a4c31ed1ac790cfff32e606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/175b3adeda8668f82a4c31ed1ac790cfff32e606"}]},"branch":"refs/heads/master"},"bf50ca502a2ea139450cc94932434e324b945f49":{"kind":"REWORK","_number":4,"created":"2025-02-10 15:21:27.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/54/941054/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/54/941054/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/4"}}},"commit":{"parents":[{"commit":"26bf3b501443ad9592446aac702afb4e6ca0b8b6","subject":"Remove reboot from basic baremetal ops","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/26bf3b501443ad9592446aac702afb4e6ca0b8b6"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-08 18:01:59.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-10 15:21:27.000000000","tz":0},"subject":"Improve check_vm_connectivity","message":"Improve check_vm_connectivity\n\nIt may take some time for cloud init to setup ssh keys,\nso its possible server start responding over ping, but SSH\nis still not awailable. This patch adds a check that we can\npass SSH authentication before doing final vm connectivity\ncheck.\n\nChange-Id: I415aff416a07a7b69baf8b152a37c3602d9702bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/bf50ca502a2ea139450cc94932434e324b945f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/bf50ca502a2ea139450cc94932434e324b945f49"}]},"branch":"refs/heads/master"},"74a70ceb80bd54a784999f0a2c2bd4a5c893700d":{"kind":"REWORK","_number":5,"created":"2025-02-10 15:43:47.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/54/941054/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/54/941054/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/5"}}},"commit":{"parents":[{"commit":"26bf3b501443ad9592446aac702afb4e6ca0b8b6","subject":"Remove reboot from basic baremetal ops","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/26bf3b501443ad9592446aac702afb4e6ca0b8b6"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-08 18:01:59.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-10 15:43:46.000000000","tz":0},"subject":"Improve check_vm_connectivity","message":"Improve check_vm_connectivity\n\nIt may take some time for cloud init to setup ssh keys,\nso its possible server start responding over ping, but SSH\nis still not awailable. This patch adds a check that we can\npass SSH authentication before doing final vm connectivity\ncheck.\n\nChange-Id: I415aff416a07a7b69baf8b152a37c3602d9702bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/74a70ceb80bd54a784999f0a2c2bd4a5c893700d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/74a70ceb80bd54a784999f0a2c2bd4a5c893700d"}]},"branch":"refs/heads/master"},"2db03082e8e1cc7733c6eaf8201b07d1d809c2f7":{"kind":"REWORK","_number":6,"created":"2025-02-10 16:13:00.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/54/941054/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/54/941054/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/6"}}},"commit":{"parents":[{"commit":"26bf3b501443ad9592446aac702afb4e6ca0b8b6","subject":"Remove reboot from basic baremetal ops","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/26bf3b501443ad9592446aac702afb4e6ca0b8b6"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-08 18:01:59.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-10 16:13:00.000000000","tz":0},"subject":"Improve check_vm_connectivity","message":"Improve check_vm_connectivity\n\nIt may take some time for cloud init to setup ssh keys,\nso its possible server start responding over ping, but SSH\nis still not awailable. This patch adds a check that we can\npass SSH authentication before doing final vm connectivity\ncheck.\n\nChange-Id: I415aff416a07a7b69baf8b152a37c3602d9702bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/2db03082e8e1cc7733c6eaf8201b07d1d809c2f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/2db03082e8e1cc7733c6eaf8201b07d1d809c2f7"}]},"branch":"refs/heads/master"},"514d3947d35a5d86f40ed3997ec2a680d2bbf9bd":{"kind":"REWORK","_number":7,"created":"2025-02-10 17:25:04.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/54/941054/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/54/941054/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/54/941054/7"}}},"commit":{"parents":[{"commit":"26bf3b501443ad9592446aac702afb4e6ca0b8b6","subject":"Remove reboot from basic baremetal ops","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/26bf3b501443ad9592446aac702afb4e6ca0b8b6"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-08 18:01:59.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-02-10 17:25:04.000000000","tz":0},"subject":"Improve check_vm_connectivity","message":"Improve check_vm_connectivity\n\nIt may take some time for cloud init to setup ssh keys,\nso its possible server start responding over ping, but SSH\nis still not awailable. This patch adds a check that we can\npass SSH authentication before doing final vm connectivity\ncheck.\n\nChange-Id: I415aff416a07a7b69baf8b152a37c3602d9702bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/514d3947d35a5d86f40ed3997ec2a680d2bbf9bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/514d3947d35a5d86f40ed3997ec2a680d2bbf9bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
