)]}'
{"id":"openstack%2Fironic-tempest-plugin~951761","triplet_id":"openstack%2Fironic-tempest-plugin~master~I3f6de02acee8d8c3764d3b1465b92292be3b690c","project":"openstack/ironic-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"36770":{"account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"last_update":"2025-07-08 13:02:43.000000000","reason":"\u003cGERRIT_ACCOUNT_36770\u003e replied on the change","reason_account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}},"37924":{"account":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"last_update":"2025-07-21 15:38:27.000000000","reason":"Change was submitted"},"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2025-07-21 13:10:41.000000000","reason":"\u003cGERRIT_ACCOUNT_24245\u003e replied on the change","reason_account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-07-21 14:02:13.000000000","reason":"\u003cGERRIT_ACCOUNT_11655\u003e replied on the change","reason_account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}},"hashtags":["ironic-week-prio"],"change_id":"I3f6de02acee8d8c3764d3b1465b92292be3b690c","subject":"Add Tempest API tests for inspection rules","status":"MERGED","created":"2025-06-04 12:38:43.000000000","updated":"2025-07-21 15:39:24.000000000","submitted":"2025-07-21 15:38:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":51,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951761","meta_rev_id":"729c07ca56fa483478f8ae9187fa4e0a3b7aa191","_number":951761,"virtual_id_number":951761,"owner":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-21 15:38:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"all":[{"value":2,"date":"2025-07-08 13:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},{"value":1,"date":"2025-07-21 13:10:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2025-07-21 14:02:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2025-07-21 14:02:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-04 12:38:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-04 13:25:45.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2025-06-16 14:35:42.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"CC"},{"updated":"2025-06-19 17:02:51.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"},{"updated":"2025-07-04 12:18:46.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"}],"messages":[{"id":"414b6eca2316106e776b1a768713cbd6b24d228b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-04 12:38:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"156d76e71c97b1ae7244ac16e15e2e95f617f533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 12:38:54.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/ironic-tempest-plugin on branch master.  The\nproblem was:\n\n  expected a list for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - publish-openstack-docs-pti\n        - tempest-plugin-jobs\n      check:\n        jobs:\n  #         # NOTE(dtantsur): keep N-3 and older non-voting for these jobs.\n  #         - ironic-tempest-functional-python3\n  ...\n\n  in \"openstack/ironic-tempest-plugin/zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":1},{"id":"1c13249c6976c145417893df54a8994dd35987b4","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-04 12:39:26.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"811ec9ec4fd425c05a9be74965314424786a1178","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-06-04 13:25:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(13 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"19bf1a535fbc27e29f2dbc6a2bf62f86d7b26672","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-04 13:55:56.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"042ac4e43c00d5a62c015fbcd0dc565004d25756","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-06-04 14:20:39.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"a1a4265539b9d2a7d7833a8d50bbf6644c9fb878","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-06-04 14:20:50.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"df3579f480085028e9d12a5690ba7e09e8b9bac0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-05 10:38:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"438516706bb3315162db17544bbfec37d0d114a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 11:10:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10ee96af2a114906a743f9ea71fb1843\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17fa772ec6504b159a22a49d769d86b3 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad02660a1bed4f51a9e14ab0b00f9cad : SUCCESS in 6m 01s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/64f1d6449e704fb4a7eba1a77a5a16d0 : FAILURE in 27m 09s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/97039caa507c4ff9b4eed79e2c784753 : FAILURE in 29m 15s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/e6a0d5594d214512a3757dbdb57f40ca : FAILURE in 21m 51s","accounts_in_message":[],"_revision_number":2},{"id":"1a2882066b4579f8faca083817370106274cb48a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-05 11:26:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d41f9474ddd1e3824887210109f76aa893e76b36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 12:01:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64adaad1d5cc43a4b836d06592bff3ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1c319cef4cf419eb234ab0009557dd0 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b020312574a34aa780400b1c916875cb : SUCCESS in 6m 01s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/50085fb9bd554e369cd86f8a46d119d5 : FAILURE in 27m 58s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/4398c2abca18447ab9fb2f03448bf519 : FAILURE in 32m 53s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/4d3f84748c1c48c681244fe03b7ac904 : FAILURE in 26m 55s","accounts_in_message":[],"_revision_number":3},{"id":"114b790aa2b28df336adb5072d756e78d66a84bf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-05 12:13:38.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"744f2df8262ab2d35c4b5973d20dd0b72371608b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 12:56:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d28197e6ace1471f8c9194e6a29272f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aba2360a13e0470a98d86cebc4309055 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f54156b0070d447987ce2765e3091fe9 : SUCCESS in 3m 47s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/63837db31a7045c3a2f7a322ad4b8c8f : FAILURE in 37m 40s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/6d82c7f2c4034f6ab5884ca489226dbb : POST_FAILURE in 9m 35s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/9e7580989855495f8f013cdcc7f77b4f : FAILURE in 36m 14s","accounts_in_message":[],"_revision_number":4},{"id":"6ccd4da8ee54435efcd7c71d3ac1a3f0d0b58a9d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-05 14:15:59.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"0ebda460c18dc65c5a9bf50a7b4cbe3a0c89dea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 14:49:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c57b5fa68e441bfa0945bdcadeed748\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f224bf84e1d4386a78c5b0f29a9525a : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cae132180bab49c2803c8467776b2434 : SUCCESS in 6m 22s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b06f57f2c99b4355be0701d9f81c9465 : FAILURE in 32m 24s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/2e140db18cdc4b63a2bcbc5d2ff4b0bd : FAILURE in 33m 02s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/250ade65f2634dbca8844e30dd892a82 : FAILURE in 26m 13s","accounts_in_message":[],"_revision_number":5},{"id":"7bbed5eba71844265239b960c755e223aac7d6d8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-05 15:04:16.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"af8de0020ac9b2bfb2437af4b66a51c416a737dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 15:49:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a6aacaa6e1347879a9905a4b937f23d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/247e9977228c46ba815ac66570f0789e : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7902a5d7d7524e318ceb746b3aafcbf8 : SUCCESS in 2m 09s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/9501b00f74324fcda9df0ec3eeac7add : FAILURE in 40m 16s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/6102a4a0c0bd4e02bf0ab70047cab2b3 : FAILURE in 38m 15s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/7f8a98a64bec4e5698aa18b02957c84d : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":6},{"id":"88e690948bbe66c07e14e0829d23d931eef8f1f7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-05 20:49:41.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"0010bb5108d84cced7d4100364f503a313113af0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 21:31:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa0b2be541b8477ea9a4ef3513125af5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea4a958bd494337923a1c7f6756c029 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c683435b04f4f6cbbd41d9d76a1d6cd : SUCCESS in 3m 20s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8762f0660ad7402faf73234b3087fe47 : FAILURE in 36m 02s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/eedd3087c643470a9950e97624a51e40 : FAILURE in 28m 45s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/c3db8009b05b40ff8f5fad1888493a12 : SUCCESS in 26m 26s","accounts_in_message":[],"_revision_number":7},{"id":"8d4943b7ddb18e23f677b9c7ec37ba4ffb60880b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-06-05 21:42:58.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"2a6a62f0f6ee9422de01ab48d53543f1de144865","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-06 14:30:29.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d9bb507c82bcff32ce3ac1cbb6201ed807c8fe91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 15:07:31.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bcef66d8c2964d2488ba389a2514a7ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8528dbc9e5a84c0b98da9f01dafe5a7e : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8f6867974f14936868f17a534262096 : SUCCESS in 4m 12s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/5a076b1c00504c1990b19ecee1d52493 : FAILURE in 36m 28s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/2f629c6614c64d9e9389a2060695e9cf : FAILURE in 29m 06s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/aa64b052dd754fe6b3c8b4736ad95c5c : SUCCESS in 23m 46s","accounts_in_message":[],"_revision_number":8},{"id":"8506adb85306ccd9f0bae79989c2603a4200f15f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-06 15:45:06.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"3df49b224dacd5742541137d4d541fe707094d74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 16:22:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28d3e29abb5d4cd690c98c2eefdb9fb6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f710811897ae47f9a2de763e58eb523d : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd606af443c34b89bb522a07e46a5318 : SUCCESS in 3m 54s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0e07de3e0d06498e80ff8ddbb83a5471 : SUCCESS in 29m 45s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/a018eae4bf694d00a838ae54abe7b869 : SUCCESS in 35m 34s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/659b6be6a6674fdfa3a1dbeb457ccdf6 : SUCCESS in 30m 50s","accounts_in_message":[],"_revision_number":9},{"id":"946f5ca5d9439544fb8c565224b9ff81d64a1141","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-09 14:13:20.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"f38326e6a35c43c65dee72997993b664c9bff0c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 14:53:59.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8246324dff88441aad95ce9d23e315c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9210d04bd72464698ac2c7092dc4457 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0daed055753c4a57a0e29e1ad6c5c1e4 : SUCCESS in 3m 40s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/eb3358dc880e4b01bf0eeb2d93f1cadb : FAILURE in 15m 26s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/38e0a1e1d7eb4d808a675dc616c2dbaf : FAILURE in 36m 57s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/1d8555205f0c4430adf0233d21197c09 : SUCCESS in 31m 34s","accounts_in_message":[],"_revision_number":10},{"id":"11ad7c63b3e3ca7c222a2da93f80709f1ee40989","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-09 15:06:58.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"22d9fa4f76ae1646fb853833967b0d8cb66f8dd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 15:58:17.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d58df86655645b6a59709a8be75ed8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3afb45b803a144a387641ce6ba0c5468 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce754bb83167434ca840897207bfb9f5 : SUCCESS in 3m 42s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2ef6bbb41a7248308723ad7003052357 : FAILURE in 35m 40s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/1c143b7eb6ae494ba83cce0e1419a44b : FAILURE in 45m 48s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/43a56e6a512f4428a4d9b97d6c3bf325 : SUCCESS in 24m 03s","accounts_in_message":[],"_revision_number":11},{"id":"b4c896398ca2e55912167432d0df23fe219b1dc4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-09 19:07:01.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"baab03fe6596e3bb95e991f886ab955c00d42bdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 19:46:48.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9d442fd000045b8afb8c0706fd6f704\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bd2ff9951f24f77a35cc4c96239f73c : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d9febd16929440ab16fd63db113c2a0 : SUCCESS in 3m 08s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/bdebc42d3e134167ae02d89bc2a9e0c5 : FAILURE in 15m 46s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/d0972f47d53443418607d6807399f47b : FAILURE in 38m 08s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/2d88024b94dd40e2aba6b44c994c5e45 : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":12},{"id":"04700bd977d69acd14d7b5c4e92cec62ebb8b8fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-11 09:48:10.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"c3a93d393bb18284796e69312e81f4486097e1cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 10:29:51.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8e89f6fa02048cebc94084366f74589\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b604878cd5844f2b836023da778c45f6 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41192fe1867c41779290e5c897196f50 : SUCCESS in 5m 04s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/7bda9f88b48c4316a9b9b416f4a866c4 : SUCCESS in 37m 48s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/c2361f893f754e128ee4f82f823a600d : SUCCESS in 36m 35s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/686e52ebff694818aef709661c40e38a : SUCCESS in 29m 07s","accounts_in_message":[],"_revision_number":13},{"id":"0319f9b4354401183b0e80af6d76c4ce36d7f523","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-11 11:22:30.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"a7cbbe44ea36844eb0a55a6c9c23a4cf5538f063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 11:56:50.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cf2f0c1690744c392a4027bdbf95488\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dfec9f4f942410a8f62edbb39edce17 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f8da1d9907e4035a89a1d5c2f960451 : SUCCESS in 2m 00s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/94947a1628cd42e893c4cf17bf986f0e : FAILURE in 14m 35s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/93abbf48336b468ea6a650d4fed1e780 : FAILURE in 33m 14s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/7a437d3780be4187b29ec7946a6c9fef : SUCCESS in 29m 41s","accounts_in_message":[],"_revision_number":14},{"id":"d51de2598df86ddcb5269e5e5d74dcb244822fb5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-11 12:33:13.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"28473e122616072a961082404f220a4c2d73d534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 13:21:17.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a9f720a0c284f0eaa74f973f3f00096\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac3b95385b54af385367f49ee1545a4 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e84d6e72e37f44e08552af69a4f0fdd4 : SUCCESS in 3m 08s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/25c1de0e3b9848babbdfdb320002de0a : FAILURE in 40m 48s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/1539b48bf2b04d808a6174c38c772f83 : FAILURE in 32m 48s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/7b131fb73b1c4e9eb91eb60e582b02e6 : SUCCESS in 26m 13s","accounts_in_message":[],"_revision_number":15},{"id":"0c0f0d10c08cefabea67eefea019f50cf65300dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-11 14:30:34.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"c7d641da2c5e8a20e971277e528bb79fc85fd17c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 15:05:32.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90c5757e6d504a77b7ca16993db301d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77233454302f4dbb95dfce2d0c26d9d4 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90ab779676cd4b51b59757e5b5311e0c : SUCCESS in 5m 20s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/b8b2c3e1dce74cbb8a2f95f076dd0527 : FAILURE in 32m 43s","accounts_in_message":[],"_revision_number":16},{"id":"207f31952b1b07a79692dce99f4e6e94d1f3e0a8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-16 14:29:09.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"a7f04863ad372ecf3c1aadc197c52ffeb5828bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 14:58:48.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffe713b246e54b76951938e84950b42c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d69dd17241405c82bc07bbe0e31ac6 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a27d4e1e3bf04bd19918e8122b2c7dca : SUCCESS in 4m 42s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/1f5263190ceb4c4e8915cc73bb2addf3 : FAILURE in 28m 14s","accounts_in_message":[],"_revision_number":17},{"id":"6de18f7e5f8c3256151ccfb3161a10ca54d68442","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-17 10:56:13.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"6a2f82873cd8a89a7fb263e9c050b4ed3a601614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 11:29:07.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab2ef6eae04147968aa1cf5a3bd2c03a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c3f45d7bce4d69ba3aea629ce731e0 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf11378fad74b5188fe1b93bf7ccdd4 : SUCCESS in 3m 32s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/94b5abf668bb4f05a5cf34807c1eef23 : FAILURE in 31m 01s","accounts_in_message":[],"_revision_number":18},{"id":"5054d2c76d97e076bd2960b75a3b88184f26013f","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-06-17 13:23:52.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"b5e0752b483af0f982d8368a67588db1931488af","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-17 13:24:57.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"825a77f12b288f240263d1f28303432381e1a5e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 14:05:29.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bff063c5c6034667a58335d8fb198693\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da886c8e1b0f490f97e12ad70c41f41f : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae653527c0a44aba8d753e196d0779c : SUCCESS in 4m 16s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ecb5562de1484456b63ef755a543b5ea : FAILURE in 38m 33s","accounts_in_message":[],"_revision_number":19},{"id":"1aa93fd296810248ed87137622ffd2db9c8e5dc2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-18 08:44:42.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"19641391ebdde4d81f060591919eb529611969a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 09:33:51.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd63afcbdea34c01bd5b625a00a10429\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e76af6fe229347edae95ad092cf1f546 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2989f66259f47a1a674abd8c9618fa8 : SUCCESS in 5m 34s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ba7f4d30c7d34e0ab124aa7932268010 : FAILURE in 47m 40s","accounts_in_message":[],"_revision_number":20},{"id":"be8d782824d0a83ca7159f53bb7d3324407edb4b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-18 11:46:03.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"6439fea0e917f5db7687eeb1513bb2ea6fa6acd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 12:11:40.000000000","message":"Patch Set 21: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c94664b12d54940b028e551be81a999\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309a688004eb41c1850309a6f108a646 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3873fab9f82441b99bb278cb5e87c211 : FAILURE in 3m 44s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/4cebafff12ca4c12b96c79a84e630ffc : FAILURE in 24m 32s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file ironic_tempest_plugin/tests/api/admin/test_inspection_rules_negatives.py has only 67 lines\u0027","accounts_in_message":[],"_revision_number":21},{"id":"3488c5d37639c300f1c80c339b37d8258c9f3d10","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-18 12:23:02.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"2e40bde5e6fb06bd51203b4a7281feadc612d620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 13:20:37.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47317b6f97d84869a6c1316726ad3561\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/496bdb3d17f54068849e2f246c8caeee : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d29cf5039094ffeaef2078eab5a2c41 : SUCCESS in 3m 27s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/768b6dd45c0f4d2083c6606ce24b9081 : FAILURE in 51m 00s","accounts_in_message":[],"_revision_number":22},{"id":"5eed8d49ccfde9abd9a641c76701096dcd6ca64e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-18 13:36:54.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"0a58c1403772729a0bd735e160d17d3c91569c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 14:17:34.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3ffe279165144018fad7bde91442ce0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a3a9b32a0045deb2b4d0a30eea9e01 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3deb40d560654c698ae9fd8928a5076c : SUCCESS in 4m 28s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/e601093ed05d4e17bcb3bd4e783544e9 : SUCCESS in 39m 49s","accounts_in_message":[],"_revision_number":23},{"id":"048f72a1cc7226ac04159b54c708e62c61a78bdb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-18 15:11:49.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"0f2ba417190b13ab648cd895a9e09e713839d1f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 15:56:31.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6715931ffdfb4f7986dbae5ca3429bd4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d850c81a034216a77fc7352caf7781 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49b5c70fa35416d95407b7701bdf860 : SUCCESS in 5m 47s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0a1f2e1b4bf5466787ff67cdcc65b479 : SUCCESS in 43m 18s","accounts_in_message":[],"_revision_number":24},{"id":"853c3d1401d87531297c7da3ee29fdd5461122f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-18 21:55:46.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"a6efbff025719fe1100eb485ef7c57210b419585","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-18 21:56:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":25},{"id":"80ddce9a08c8cbe5dc258f9e30d195c7ef8d1654","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-18 21:57:15.000000000","message":"Patch Set 25:\n\n(16 comments)","accounts_in_message":[],"_revision_number":25},{"id":"b2102163d17be43c9a3909a8849836aac20a76ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 00:49:00.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5817dea9cc649af91879eaf582dae70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb09f2c77ef94d7ea03ab5c512f416c8 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df96d6465f444b11b6837f4bbbf9d6bc : SUCCESS in 4m 18s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/990a7b4e5ddb40a1a7e71f55b9ef2580 : SUCCESS in 46m 14s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/4eabfe8a02f448b19967242147d59fb7 : FAILURE in 40m 58s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/bef9cb4f4dac40d687820425b1dfa8a5 : SUCCESS in 32m 51s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/964a4b1ecb664ad9aa811bf3270b0447 : SUCCESS in 39m 09s\n- ironic-standalone-anaconda-2025.1 https://zuul.opendev.org/t/openstack/build/2a5f094650404399bf919ea8eb89526f : SUCCESS in 38m 23s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/57495568b4314ae380e3c1257f84db5d : SUCCESS in 44m 29s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/901b775bbd4a47edb983e2043a0cf931 : SUCCESS in 51m 54s\n- ironic-standalone-redfish-2025.1 https://zuul.opendev.org/t/openstack/build/821ad3d682234f368a3dd71b9b58333b : SUCCESS in 1h 19m 37s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/970afb9f00a74f7fafb7121ae76812bb : SUCCESS in 1h 26m 06s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/7662a6032dfd4cb19d7d4f105c2e2e74 : SUCCESS in 49m 59s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/b252bd697c104c4fb271139fce7194ed : SUCCESS in 2h 48m 08s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2025.1 https://zuul.opendev.org/t/openstack/build/01f748e7aeaa47c087726daa7b25c951 : SUCCESS in 2h 29m 25s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/d378aacb1691482493b3ae98b1972502 : SUCCESS in 47m 13s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"e7945ba86a22af5e2df093b280b1eb3b80bfd44f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-19 08:29:15.000000000","message":"Uploaded patch set 26: New patch set was added with same tree, parent tree, and commit message as Patch Set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"4b1cbf1ace15614e17bb69f22453cb7627f2569f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 10:58:56.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/149d18ae4f694badb1e6db04fa734677\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc03dc38d51435d854e673bbbea27bd : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b089f7b2135a497f8f4353dd3f3460f3 : SUCCESS in 4m 31s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/32a12547d6fd453d8c08e08380ddf49a : SUCCESS in 52m 28s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/6feb89d6f52645ec859f986bcca509af : SUCCESS in 29m 19s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/f6e5ec8d03cd448dac6794c59a2b4c57 : SUCCESS in 24m 39s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/51dded593337417ba4c5bc796d3239c4 : SUCCESS in 39m 41s\n- ironic-standalone-anaconda-2025.1 https://zuul.opendev.org/t/openstack/build/43c2648eeef84b4c8a3b012a7bab4ae5 : SUCCESS in 36m 56s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/8c5e46da51e649e8adc9d3751662a994 : SUCCESS in 38m 11s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/19db13d3f51842bf9680470335d09d7f : SUCCESS in 53m 18s\n- ironic-standalone-redfish-2025.1 https://zuul.opendev.org/t/openstack/build/bdf1918345eb4f6f9484fe23fc33491f : SUCCESS in 55m 27s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/191ab35631da494cbf101b1c2d26bafd : SUCCESS in 1h 05m 49s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/55f3529ecc7544408827a093beef68c4 : SUCCESS in 38m 08s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/dac866ca6db948819102cdc7a8bebe7c : SUCCESS in 1h 49m 40s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2025.1 https://zuul.opendev.org/t/openstack/build/2ae46e047b1046c3bf93a4da9fa48b69 : SUCCESS in 2h 10m 18s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/10efb06734694fe59053d70ece483b1b : SUCCESS in 41m 05s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"85c6ecb961fd3a088e87cd3edb8252929c1b45ba","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-06-19 17:02:51.000000000","message":"Patch Set 26: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"f4bdce607f518139af7e4dde5cc4bf35a30f35f7","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-06-23 13:12:36.000000000","message":"Patch Set 26: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":26},{"id":"73e10c690158128329fa181b453edc6149126b53","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-23 14:12:32.000000000","message":"Patch Set 26: Workflow-1","accounts_in_message":[],"_revision_number":26},{"id":"e628b6840c80f9902a87aa2f9b8c9b0ca86c0f95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-24 14:26:41.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":27},{"id":"fac160369bd2e425b5b9047246a7885496c11493","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-24 14:27:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":27},{"id":"f7116d95687d2c23737bcbe0cc4a6eb640d5829a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 15:13:12.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b316a407f099439f9a38502bd777fc7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c0f32b16ce84da6a550c95034e3b95a : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a32ac24b2936427097c047f07027a173 : SUCCESS in 3m 43s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0a4ed4446a4e47d1bbe5ab50d51ce2e4 : FAILURE in 45m 23s","accounts_in_message":[],"_revision_number":27},{"id":"5840b0ad6fd2d3477c03294cf353526b6af4ec49","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-25 08:21:41.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"814023eff8caf663742668cdd76fcee8c77f94fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 09:12:21.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d14524248cd40f6967ec23149da7eec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/decad1c621ba4c20a45d0af232228976 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6a3dd244b11435b9686a1c23a61d623 : SUCCESS in 3m 27s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/d48df328a9dd40aaae25a887b77a2b66 : FAILURE in 50m 02s","accounts_in_message":[],"_revision_number":28},{"id":"3ff520df1c62493fa7abef03d3612a57fb86d1ed","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-25 09:50:53.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"377911969fe8fba593bf340057d513a3287bb28e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 10:53:44.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6157a71708d3413b9bd66b0735e40544\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0d6019d06743cf8db0897c2d8ad45d : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b312e5b1fa95482da81ff7f57b69f0a2 : SUCCESS in 5m 39s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/fc2b3918064744ceb7b4e2a26ede9e2b : FAILURE in 42m 43s","accounts_in_message":[],"_revision_number":29},{"id":"93fe0f193fd6fedc52d1da6a153e6bfa19161390","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-25 11:07:32.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"18afe9e3af32c31a39aa730a1c13009c23ba9c0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 12:00:15.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6a1c9bbbfa74eefa3cab341d2e104d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f871d47ed5b4e99a46fd0970526af63 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed87706675f4b489dc0e8c902512f5c : SUCCESS in 2m 42s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/37bc1b32c4d448d898cf2b4126a72f91 : FAILURE in 47m 10s","accounts_in_message":[],"_revision_number":30},{"id":"0b6e191936267e9b246678d184b0b3157d873c8a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-06-30 13:46:50.000000000","message":"Uploaded patch set 31: New patch set was added with same tree, parent tree, and commit message as Patch Set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"e406c454decaba9cfe8c4229554f9f776142397f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 14:53:41.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3837529337748968090f9c970fdbdc0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe481b4481d43908a44c7ba59c34179 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d5cb378e8b4de69e35139aee25d65a : SUCCESS in 4m 05s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c21c23cf6e4c4fb3bf79f396c97a0180 : SUCCESS in 47m 13s","accounts_in_message":[],"_revision_number":31},{"id":"9acb366d80a777dd07fa04214f01565521db9abc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-02 08:57:32.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"7bf1d1582dffa23ad990053555a0eea22449d65b","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-02 09:02:06.000000000","message":"Patch Set 32:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":32},{"id":"b1d37fffad50a62ef973857ff99f4d26b26bed21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 11:28:39.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1008ec0d013d4250800882dbba177de3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b65d228eadba4f858da705afde4b829a : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94fdacd7358b4d19a51b5af4180d3878 : SUCCESS in 3m 16s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/f6282744a8dc481abc3a2f9c616a9ee9 : SUCCESS in 48m 52s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/f9669b52799a435d873cb1c7924ce018 : POST_FAILURE in 9m 55s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/426e088437b446da90fe78baebf2ed1c : SUCCESS in 29m 03s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/69349627df6846a6a7511ea8cfb214e7 : SUCCESS in 51m 54s\n- ironic-standalone-anaconda-2025.1 https://zuul.opendev.org/t/openstack/build/7957b94aca0b4cc2a5f45a79a41cd795 : SUCCESS in 45m 01s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/c1538a62fc9d4119b0095bbe2b4f8219 : SUCCESS in 17m 19s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/88441a6ad27b4e46843448d0acc056e4 : SUCCESS in 1h 05m 29s\n- ironic-standalone-redfish-2025.1 https://zuul.opendev.org/t/openstack/build/f87da9d584b744cc92066527b6e51f38 : SUCCESS in 1h 18m 17s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/e16e5eb68f70402d9fe01f17fd3a6398 : SUCCESS in 1h 14m 34s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3436f920c375460892c67571677c2ce5 : SUCCESS in 19m 50s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/af81378a9720452c8335932018131635 : FAILURE in 59m 28s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2025.1 https://zuul.opendev.org/t/openstack/build/89b324eb7c1c4ad896b9633acb18388b : FAILURE in 2h 18m 35s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/d5ef2c437dd741bfaf17789186b51b21 : SUCCESS in 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"82213e55259917533ff0d57fdbc1303aff3dbb9e","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-02 12:29:12.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"37ee0c3c491da693e513c36e127f532311e8c9c3","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-02 16:06:24.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"4df6bacf882ef4ac0abf24094de529cd77fbb97b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 16:20:35.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2122231abeda437a8df77e14fb0b1aa0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e253f2f71c94491ca952f647b2afead5 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/923b073ee9d6454d815432344d8fb39c : SUCCESS in 3m 38s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/d8529f52e09f4648b8c2b6d88e922569 : SUCCESS in 48m 07s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/4ceaf82a572d4d1980084dee1ba7b83e : SUCCESS in 30m 57s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/53895781a0424aeea33808ec31e9f2a5 : SUCCESS in 29m 24s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/147f81cfee0c4d318a218dc70c2db5fc : SUCCESS in 50m 26s\n- ironic-standalone-anaconda-2025.1 https://zuul.opendev.org/t/openstack/build/a32d67eeab8b4705bce5feeef23198d6 : SUCCESS in 42m 18s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/695218e34c454c528fddcf3b257bf3c7 : SUCCESS in 42m 45s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/ffe87d04c3a74d799bf03fab34dbbc87 : SUCCESS in 51m 02s\n- ironic-standalone-redfish-2025.1 https://zuul.opendev.org/t/openstack/build/18571fd5ade34cdc93e17654ec82fa23 : SUCCESS in 1h 23m 49s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/28feb67fd02c47cbaa04cf7237c124e9 : SUCCESS in 1h 19m 13s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/6f1ae510efa44fcf8e79a836955d4779 : SUCCESS in 52m 14s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/03aad710f1b64f7e9f45b6c137c6bcd8 : FAILURE in 1h 07m 18s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2025.1 https://zuul.opendev.org/t/openstack/build/79339d2a4135458686ab2acc821896e3 : FAILURE in 2h 22m 50s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/f5f180664f7a4d69be4b8d4c611d8c44 : SUCCESS in 45m 40s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"69aa78cf22f8c08b9e9dc0a040787b3d868831f9","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-03 03:31:04.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"075abe2a781061e4b62723870be042a906ab4d14","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-07-03 12:31:00.000000000","message":"Patch Set 32: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"18af88fbbdc0639f5e34aa1dcde03fd93868bc79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-03 12:42:18.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"cc2b790c5746e70c17bafcbcdef0fcfa6189b914","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-03 12:45:46.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"8dd8f1f5ef430ea042145e39c3f57313359f8c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 15:34:17.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/913f428b7dc34dc8805eb4a2c2c275d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a495e4726334a96b7dea021b12b0074 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/485eb8e25b864d12b009f0d55a498da4 : SUCCESS in 2m 33s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/5819f5df5fa844d5b7442b45b761ee21 : SUCCESS in 31m 29s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/3a3a249e288a4c788085bfd4be0d5467 : FAILURE in 34m 00s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/4d6009d2dcdb4e04a6ae33ac68476dc0 : SUCCESS in 17m 02s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/5e8def10bb1c49ffaaa5205ce3250653 : SUCCESS in 38m 01s\n- ironic-standalone-anaconda-2025.1 https://zuul.opendev.org/t/openstack/build/9fa0500dc3324f1aa78c5a92d487e829 : SUCCESS in 47m 08s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/02a8090bd5b1406baef1b4adfb1ff01b : SUCCESS in 18m 44s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/fc081d8eacee4a4dbc04186f4fb4688c : SUCCESS in 1h 08m 41s\n- ironic-standalone-redfish-2025.1 https://zuul.opendev.org/t/openstack/build/1d87b715e414489b926cdba19c398107 : SUCCESS in 1h 10m 12s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/c6530dba134e4dbfb798bd3ace545d65 : SUCCESS in 1h 10m 28s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/585e90676baa4ac9aea644518835ea15 : SUCCESS in 42m 18s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/308c0d91f1494637907aa1b27430cca9 : SUCCESS in 2h 40m 32s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2025.1 https://zuul.opendev.org/t/openstack/build/37ec7f8e80bc4a519f057542998443aa : SUCCESS in 2h 29m 37s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/3ae4d0e7e5274d4ea47c3c814414ad48 : SUCCESS in 42m 39s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"54e1f19c1dcf2577ff12dc12b1da6b51e812fd31","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-04 09:19:05.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"e60ff8ff85ab3d13a2caf5dc104056871a02eb4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 10:48:03.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42ab441a302e4cada05828d4fd763c7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ff1c99b53c41c193fc55bba22e592d : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b721eb9316944d8a0785bfb2cea27c9 : SUCCESS in 2m 35s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/1910f2e099f14f0e8e8c5d81219ea68d : SUCCESS in 41m 17s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/94e6340a6c474f67a554dea8d5707071 : SUCCESS in 26m 49s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/e07be9509d374fe7826211f1886b7e6f : SUCCESS in 26m 22s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/f32e5b5422854cd28e3ee929e684c0ef : SUCCESS in 40m 21s\n- ironic-standalone-anaconda-2025.1 https://zuul.opendev.org/t/openstack/build/e9779ff9af214626ac9ee28a9be9c0bb : SUCCESS in 30m 40s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/85a8716a925644feaf26085237c7d5df : SUCCESS in 49m 22s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/b2dd2954474e42bd829560c63aa79a70 : SUCCESS in 51m 58s\n- ironic-standalone-redfish-2025.1 https://zuul.opendev.org/t/openstack/build/a783c34947c14791ba397a42bbdc9c79 : SUCCESS in 1h 23m 03s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/30ef6a5b263740c4b788de76819e0980 : SUCCESS in 54m 07s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/2ffcec54f1c7477f95cfb8ed0b27742e : SUCCESS in 36m 36s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/82d286c7b95d4011b0472d5d3ab1b395 : FAILURE in 1h 18m 21s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2025.1 https://zuul.opendev.org/t/openstack/build/a760ce49a4ef4473bf08f56bc31c1168 : SUCCESS in 1h 10m 17s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/a515313e39874547808eecccbe66a2a5 : SUCCESS in 28m 33s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"ad8571a7ca3593c5374a04126d2b40c53937d0f5","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-07-04 12:18:46.000000000","message":"Patch Set 33: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"e8a0c77d2fabf50e2e79b087a3318f65e3053500","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-07-07 08:58:00.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":33},{"id":"d6f6d2b0039e67bec2ddfdc48a68a58ef5fea163","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-07-07 14:02:12.000000000","message":"Patch Set 33: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"c51c9b49791f3af68fa93ec7fd10b533d1dcb036","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-07 16:08:01.000000000","message":"Uploaded patch set 34: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"4ed9a4676b5bb3d4281be6387f4a9fab27b3838e","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-07 16:08:22.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"8c0ef6601ee855c02bdc1fe726232b30ca531f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 17:57:42.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc29fad239a742348878d41b7046323c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2571e34508e44834a3914ec8665d5b2b : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9432c2fbb58e4ca5b68ed001b9de3f9c : SUCCESS in 3m 21s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/6436f6dfdeb3421cb6e884d7ded11539 : SUCCESS in 51m 16s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/f1dc3c1df3414c98a1c6ef8b338f69ec : POST_FAILURE in 9m 20s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/f700df372e83422a918f46b3b0db7d4c : SUCCESS in 30m 33s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/b601362dd3894d8995214e8dc2dfa712 : SUCCESS in 20m 33s\n- ironic-standalone-anaconda-2025.1 https://zuul.opendev.org/t/openstack/build/0b0f7ec23f1d4b3d98fc1d921dd4b932 : SUCCESS in 40m 06s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/a78efd3d2ace4e44af1205178bdd99f9 : SUCCESS in 17m 31s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/67e13e3815df4d338fc2ed5dfaf81cb8 : SUCCESS in 51m 24s\n- ironic-standalone-redfish-2025.1 https://zuul.opendev.org/t/openstack/build/f89f3cb935c245b490702fc323da2312 : SUCCESS in 44m 06s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/50e0842d592d4bbebc37dfd22f29311e : SUCCESS in 1h 31m 57s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/837e3a10581e40c090ec4576d6dcd290 : SUCCESS in 47m 09s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/dc2899532b424b9d9343bdcc3227ef19 : FAILURE in 54m 02s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2025.1 https://zuul.opendev.org/t/openstack/build/3bdf4b919c4d4dfabd0e1702fd4aec25 : SUCCESS in 1h 40m 05s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/4489b70342bd411db12363499c3197fb : SUCCESS in 40m 30s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"ad3c6c5e50d67dc95e12e141ae0e4be92292c379","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-07 19:02:57.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"fb31080479df86ed823d10ee0410bd05dfc62ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 20:55:44.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c81b04ccfaa54b71844daad14ebc2124\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be364b24301e4346b17ea8d071156c6c : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13813a0991254c41bfe3f2c953a776e2 : SUCCESS in 4m 45s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2e0c492974e04c269fbe2d7bb58f426a : SUCCESS in 51m 55s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/11ec70e5fc594d09abd064b3326f9a9d : SUCCESS in 43m 41s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/c6107a7bf02b4e0493da4d17901a773e : SUCCESS in 23m 02s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/f483ed5c00f7448aad96b98cb70c670c : SUCCESS in 43m 31s\n- ironic-standalone-anaconda-2025.1 https://zuul.opendev.org/t/openstack/build/6c4f1d188f854cb6b13f4cee6181dfd8 : SUCCESS in 54m 33s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/8e2f3fe79c654920a24ad04622fd934e : SUCCESS in 26m 15s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/ba675f0120b94d88b296ef6d4cd7ebd7 : FAILURE in 1h 02m 13s\n- ironic-standalone-redfish-2025.1 https://zuul.opendev.org/t/openstack/build/fdab738dc92a41f9abfa37187a7ed0e2 : SUCCESS in 1h 32m 04s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/100af5dd5ccf42c58426f52912683289 : SUCCESS in 44m 50s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/cdadd178b4194d77b4b05be435f2a2fb : SUCCESS in 23m 25s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/8de356b35a1f4535bd4388fdbfa94257 : SUCCESS in 1h 50m 38s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2025.1 https://zuul.opendev.org/t/openstack/build/42a111b946da4cdeba18e92ab9c75199 : SUCCESS in 1h 48m 45s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/db5ef3d0507a434d9fa8e3a658bd15e8 : POST_FAILURE in 17m 28s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"707c80909015f1918dbabd0c8fe4ff03b7fc4e38","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-07-08 13:02:43.000000000","message":"Patch Set 34: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":34},{"id":"4ec272aeb63c462268fbda8dd0033017113af479","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-07-08 13:02:53.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"57395047fe90f446b239822c47e2f43299a8f11c","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-08 14:11:07.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"7659ba648236776e320354f6637a037f9a854323","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 14:39:21.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cf9d43192e44eacb9e7faa6153dc7e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a401bf401e704c04959b67c276e28556 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89030c9a26204d348a0a40809a5a5896 : SUCCESS in 3m 53s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/6d47a0053bbc46e88dc2460353f21797 : SUCCESS in 41m 25s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/dff0094a11574b838b6f43c87f16e3a9 : SUCCESS in 26m 47s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/26575a076d0e40bba62a18fccc7c796b : SUCCESS in 21m 14s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/a9799afba3644864b4a1ff0f21c7f7d8 : SUCCESS in 1h 10m 53s\n- ironic-standalone-anaconda-2025.1 https://zuul.opendev.org/t/openstack/build/0f13d7c47c2c4fb1b8ad617d60d44d4e : SUCCESS in 21m 02s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/e0163822e7974d52b8bd160d88a8fec5 : SUCCESS in 16m 24s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/676340ca2da44c10b80af1ae5c18ece5 : FAILURE in 48m 42s\n- ironic-standalone-redfish-2025.1 https://zuul.opendev.org/t/openstack/build/86a056e0e25e490dba5bfb766e411901 : SUCCESS in 37m 27s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/375f23f1e3664e0e9a9634d9606e012c : SUCCESS in 34m 01s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/493a52099b614ff59be44b92a25cc1bb : SUCCESS in 19m 34s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/408bec267747451e9df6307edf014ba6 : SUCCESS in 1h 33m 26s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2025.1 https://zuul.opendev.org/t/openstack/build/00aeb50497ec4402b030d90a97699436 : SUCCESS in 1h 13m 27s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/09a3e22f640d4f96b69c51497a12af7c : SUCCESS in 21m 07s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"a87db4ec889a99d0c8e8b410ba96fd82ebda060d","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-07-10 14:28:32.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"8b6ef2d84b24065bfd945a6106a8aa1474229c3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:47:36.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/112a051987164fc99f821618130e0793\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca63493340924996b2a1942ef7736895 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db92d016a4c64ab5bf7b498c916c26c9 : SUCCESS in 6m 14s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/16be1a618c8b4fbc971bd2b3bcce532f : SUCCESS in 50m 49s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/5b5a5fb6d2a94782a6607d611114558f : SUCCESS in 38m 01s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/9c1eff10173947479c4e66cf69892ed9 : SUCCESS in 30m 08s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/a7082291a59c4920bb3895b266fb6d7a : SUCCESS in 51m 21s\n- ironic-standalone-anaconda-2025.1 https://zuul.opendev.org/t/openstack/build/16efbbc5768341239fec185a0dd333c9 : SUCCESS in 49m 16s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/017fa871d3074b6382be6164d12ea38a : SUCCESS in 42m 55s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/6c08789cf97b41e4bf8b15e578f35f6d : SUCCESS in 50m 18s\n- ironic-standalone-redfish-2025.1 https://zuul.opendev.org/t/openstack/build/b277fe015990451f9fbf76faa317c6d4 : SUCCESS in 1h 10m 25s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/49a6a1b0718344d78284645401b0db6b : SUCCESS in 1h 05m 19s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/c30f6272bfb44c3fa61092f0724c200b : SUCCESS in 38m 56s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/a8ac7c2e11b34bbb98f67080f76c8290 : SUCCESS in 2h 12m 40s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode-2025.1 https://zuul.opendev.org/t/openstack/build/070a7df299274c278ad78689d8b6a21e : SUCCESS in 2h 07m 19s (non-voting)\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/a8d0a1cf7cfc464a8f974ba8fd1b55f1 : SUCCESS in 43m 35s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"e8fe64c33702ee847bfc6ae6277173ef1087e5fc","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-07-21 13:10:41.000000000","message":"Patch Set 34: Code-Review+1","accounts_in_message":[],"_revision_number":34},{"id":"b20dbd999fe034559fce8a811acc3dea79b45e76","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-07-21 14:02:13.000000000","message":"Patch Set 34: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":34},{"id":"bcc561c62e04cbd94bea2c89f4410af979772c46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 14:02:50.000000000","message":"Patch Set 34: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":34},{"id":"86eeda0c05eb34f0b10c13f728bc828b3149550e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 15:38:27.000000000","message":"Patch Set 34: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/231c384ce3f8415581113780f79fa98d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72473ae29ffc4e379dd04b1196e3aca0 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ceb06e2d81433d9b319cf77508f7c0 : SUCCESS in 3m 32s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/56b0b731a1284e588468b82117978c27 : SUCCESS in 47m 39s\n- ironic-tempest-functional-python3-2025.1 https://zuul.opendev.org/t/openstack/build/ad504597cac545b0a5fae5468fd4663f : SUCCESS in 34m 47s\n- ironic-tempest-functional-python3-2024.2 https://zuul.opendev.org/t/openstack/build/2776c27304d444719746bf8de7843036 : SUCCESS in 17m 21s\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/06431534dacc43b9bb550f4aa39555d4 : SUCCESS in 36m 34s\n- ironic-standalone-anaconda-2025.1 https://zuul.opendev.org/t/openstack/build/ee3efc78077649ff92dcdba96015b5b0 : SUCCESS in 58m 04s\n- ironic-standalone-anaconda-2024.2 https://zuul.opendev.org/t/openstack/build/9cb02568f63b4a9fac29de74f6a466b4 : SUCCESS in 42m 02s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/95f59ef041a440e58a1cdbf3b9eb565c : SUCCESS in 1h 00m 58s\n- ironic-standalone-redfish-2025.1 https://zuul.opendev.org/t/openstack/build/d891cfa1548748968f803c62d9eb9c59 : SUCCESS in 1h 25m 39s\n- ironic-standalone-redfish-2024.2 https://zuul.opendev.org/t/openstack/build/d7e6fbf456db40b19d9bb1bef78ba171 : SUCCESS in 1h 28m 38s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/8ffb292e7fa4473195d3d9063a02dd6f : SUCCESS in 19m 57s","accounts_in_message":[],"_revision_number":34},{"id":"862910e5ac8d65feb463a5f3cf05f39ff9900945","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 15:38:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":34},{"id":"729c07ca56fa483478f8ae9187fa4e0a3b7aa191","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 15:39:24.000000000","message":"Patch Set 34:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f2d669fdb284c02a1fb4e0124a7e6f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c9f68ff5234981b3ccb7661348f580 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":34}],"current_revision_number":34,"current_revision":"fa630483afe3b469c2eddeb9aba57ddf3c807d66","revisions":{"4d9a4d3a21ce0db1edad17c727bfe5c3f9e56e1d":{"kind":"REWORK","_number":1,"created":"2025-06-04 12:38:43.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/1"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:38:07.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within Ironic.\n\nticket: https://bugs.launchpad.net/ironic/+bug/2105478\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/4d9a4d3a21ce0db1edad17c727bfe5c3f9e56e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/4d9a4d3a21ce0db1edad17c727bfe5c3f9e56e1d"}]},"branch":"refs/heads/master"},"c900ddc79b3154a1ebe7c9d49f28cf2f46c8b3be":{"kind":"REWORK","_number":2,"created":"2025-06-05 10:38:26.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/2"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-05 10:36:35.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/c900ddc79b3154a1ebe7c9d49f28cf2f46c8b3be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/c900ddc79b3154a1ebe7c9d49f28cf2f46c8b3be"}]},"branch":"refs/heads/master"},"829a61ea3eb442907b54fc723d6c228332b85f2a":{"kind":"REWORK","_number":3,"created":"2025-06-05 11:26:21.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/3"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-05 11:26:02.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/829a61ea3eb442907b54fc723d6c228332b85f2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/829a61ea3eb442907b54fc723d6c228332b85f2a"}]},"branch":"refs/heads/master"},"5506420f7cea9988a4779b857fb9e93d72fc4266":{"kind":"REWORK","_number":4,"created":"2025-06-05 12:13:38.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/4"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-05 12:11:48.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/5506420f7cea9988a4779b857fb9e93d72fc4266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/5506420f7cea9988a4779b857fb9e93d72fc4266"}]},"branch":"refs/heads/master"},"373ca6b425aef73dc927ef78e7278abf38400c6e":{"kind":"REWORK","_number":5,"created":"2025-06-05 14:15:59.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/5"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-05 14:15:47.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/373ca6b425aef73dc927ef78e7278abf38400c6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/373ca6b425aef73dc927ef78e7278abf38400c6e"}]},"branch":"refs/heads/master"},"05aa237fc61879d6f0964cc8dc870a1a8e70f59b":{"kind":"REWORK","_number":6,"created":"2025-06-05 15:04:16.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/6"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-05 14:59:29.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/05aa237fc61879d6f0964cc8dc870a1a8e70f59b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/05aa237fc61879d6f0964cc8dc870a1a8e70f59b"}]},"branch":"refs/heads/master"},"1fc75bdb6c22da2ff874ffb6c94cdb82820cb37b":{"kind":"REWORK","_number":7,"created":"2025-06-05 20:49:41.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/7"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-05 20:49:18.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/1fc75bdb6c22da2ff874ffb6c94cdb82820cb37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/1fc75bdb6c22da2ff874ffb6c94cdb82820cb37b"}]},"branch":"refs/heads/master"},"433a493d3c5755f1c8684510cb77d52b39f906b7":{"kind":"REWORK","_number":8,"created":"2025-06-06 14:30:29.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/8"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-06 14:30:17.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/433a493d3c5755f1c8684510cb77d52b39f906b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/433a493d3c5755f1c8684510cb77d52b39f906b7"}]},"branch":"refs/heads/master"},"d68d81fecb19c865f6243b1f870a60bca2ba4fa3":{"kind":"REWORK","_number":9,"created":"2025-06-06 15:45:06.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/9"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-06 15:44:56.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/d68d81fecb19c865f6243b1f870a60bca2ba4fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/d68d81fecb19c865f6243b1f870a60bca2ba4fa3"}]},"branch":"refs/heads/master"},"f920ae8c8f7d5c39a4f653a14406cbc5c544486d":{"kind":"REWORK","_number":10,"created":"2025-06-09 14:13:20.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/10"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-09 14:13:03.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/f920ae8c8f7d5c39a4f653a14406cbc5c544486d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/f920ae8c8f7d5c39a4f653a14406cbc5c544486d"}]},"branch":"refs/heads/master"},"06c824ecc5aa6110c13986dfe2da433b2c00f88b":{"kind":"REWORK","_number":11,"created":"2025-06-09 15:06:58.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/11"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-09 15:06:49.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/06c824ecc5aa6110c13986dfe2da433b2c00f88b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/06c824ecc5aa6110c13986dfe2da433b2c00f88b"}]},"branch":"refs/heads/master"},"1f87f2d88898e8d880a5df9477c1c6824151b3da":{"kind":"REWORK","_number":12,"created":"2025-06-09 19:07:01.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/12"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-09 19:06:50.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/1f87f2d88898e8d880a5df9477c1c6824151b3da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/1f87f2d88898e8d880a5df9477c1c6824151b3da"}]},"branch":"refs/heads/master"},"a877338f63ba13a0b5de120bb5df7afedf69ca2a":{"kind":"REWORK","_number":13,"created":"2025-06-11 09:48:10.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/13"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-11 09:48:01.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/a877338f63ba13a0b5de120bb5df7afedf69ca2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/a877338f63ba13a0b5de120bb5df7afedf69ca2a"}]},"branch":"refs/heads/master"},"593f91fcb5366ad1f98dd6bec82691372c3f8c87":{"kind":"REWORK","_number":14,"created":"2025-06-11 11:22:30.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/14"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-11 11:22:21.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/593f91fcb5366ad1f98dd6bec82691372c3f8c87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/593f91fcb5366ad1f98dd6bec82691372c3f8c87"}]},"branch":"refs/heads/master"},"dfd774d61b826c8e71cbe79db23c0f40bb7e577b":{"kind":"REWORK","_number":15,"created":"2025-06-11 12:33:13.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/15"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-11 12:33:02.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/dfd774d61b826c8e71cbe79db23c0f40bb7e577b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/dfd774d61b826c8e71cbe79db23c0f40bb7e577b"}]},"branch":"refs/heads/master"},"be9f4f94f451007e1d1d11b18a398d366f66cdba":{"kind":"REWORK","_number":16,"created":"2025-06-11 14:30:34.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/16"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-11 14:30:23.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/be9f4f94f451007e1d1d11b18a398d366f66cdba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/be9f4f94f451007e1d1d11b18a398d366f66cdba"}]},"branch":"refs/heads/master"},"b4db6ef8c256233d7b5d881125858224fc5ec699":{"kind":"REWORK","_number":17,"created":"2025-06-16 14:29:09.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/17"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-16 14:28:33.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/b4db6ef8c256233d7b5d881125858224fc5ec699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/b4db6ef8c256233d7b5d881125858224fc5ec699"}]},"branch":"refs/heads/master"},"ae77b5172fb1d49356167208de8bcdfb0831e9e2":{"kind":"REWORK","_number":18,"created":"2025-06-17 10:56:13.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/18"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-17 10:56:02.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/ae77b5172fb1d49356167208de8bcdfb0831e9e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/ae77b5172fb1d49356167208de8bcdfb0831e9e2"}]},"branch":"refs/heads/master"},"90c794eba247ec4fb16d2b5a75d451164d812140":{"kind":"REWORK","_number":19,"created":"2025-06-17 13:24:57.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/19"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-17 13:24:47.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/90c794eba247ec4fb16d2b5a75d451164d812140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/90c794eba247ec4fb16d2b5a75d451164d812140"}]},"branch":"refs/heads/master"},"d7237a3fd1b09c9dc403dff553739d24a4f34585":{"kind":"REWORK","_number":20,"created":"2025-06-18 08:44:42.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/20"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-18 08:44:25.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/d7237a3fd1b09c9dc403dff553739d24a4f34585"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/d7237a3fd1b09c9dc403dff553739d24a4f34585"}]},"branch":"refs/heads/master"},"e56363bd2833815dc2f4581c29e0782a70b95cb9":{"kind":"REWORK","_number":21,"created":"2025-06-18 11:46:03.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/21"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-18 11:45:53.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/e56363bd2833815dc2f4581c29e0782a70b95cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/e56363bd2833815dc2f4581c29e0782a70b95cb9"}]},"branch":"refs/heads/master"},"5f3af7141eb8ab3db812c7d9227967709f8416c0":{"kind":"REWORK","_number":22,"created":"2025-06-18 12:23:02.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/22"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-18 12:20:33.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/5f3af7141eb8ab3db812c7d9227967709f8416c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/5f3af7141eb8ab3db812c7d9227967709f8416c0"}]},"branch":"refs/heads/master"},"908a1e666afed8381927f84ae5dac9e5577e3705":{"kind":"REWORK","_number":23,"created":"2025-06-18 13:36:54.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/23"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-18 13:36:45.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/908a1e666afed8381927f84ae5dac9e5577e3705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/908a1e666afed8381927f84ae5dac9e5577e3705"}]},"branch":"refs/heads/master"},"df566972f78be22bd02a25622139c73debaf29ed":{"kind":"REWORK","_number":24,"created":"2025-06-18 15:11:49.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/24"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-18 15:11:40.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the new inspection\nrules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/df566972f78be22bd02a25622139c73debaf29ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/df566972f78be22bd02a25622139c73debaf29ed"}]},"branch":"refs/heads/master"},"c2c3a349e376d4ddd0c6ed053651df81da54d451":{"kind":"REWORK","_number":25,"created":"2025-06-18 21:55:46.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/25"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-18 21:54:20.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the creation of new\ninspection rules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/c2c3a349e376d4ddd0c6ed053651df81da54d451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/c2c3a349e376d4ddd0c6ed053651df81da54d451"}]},"branch":"refs/heads/master"},"33d93fc0aa394a57106a601ce6f4747672763710":{"kind":"NO_CHANGE","_number":26,"created":"2025-06-19 08:29:15.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/26"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-19 08:29:00.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the creation of new\ninspection rules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/33d93fc0aa394a57106a601ce6f4747672763710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/33d93fc0aa394a57106a601ce6f4747672763710"}]},"branch":"refs/heads/master"},"098ee5e7b3d14ebc53d4eb1a2c1136c2c5ca2ee0":{"kind":"REWORK","_number":27,"created":"2025-06-24 14:26:41.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/27"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-24 14:25:53.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the creation of new\ninspection rules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/098ee5e7b3d14ebc53d4eb1a2c1136c2c5ca2ee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/098ee5e7b3d14ebc53d4eb1a2c1136c2c5ca2ee0"}]},"branch":"refs/heads/master"},"fd69f5466e85a2756f8c6c5e7ab93f5eb20786c4":{"kind":"REWORK","_number":28,"created":"2025-06-25 08:21:41.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/28"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-25 08:21:18.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the creation of new\ninspection rules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/fd69f5466e85a2756f8c6c5e7ab93f5eb20786c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/fd69f5466e85a2756f8c6c5e7ab93f5eb20786c4"}]},"branch":"refs/heads/master"},"7ad44806248a179de7b67ae9b08b4cf2a0d8d268":{"kind":"REWORK","_number":29,"created":"2025-06-25 09:50:53.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/29"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-25 09:50:38.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the creation of new\ninspection rules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/7ad44806248a179de7b67ae9b08b4cf2a0d8d268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/7ad44806248a179de7b67ae9b08b4cf2a0d8d268"}]},"branch":"refs/heads/master"},"e59ca38a90774a9dd86be44ae893dfcfb9c44b5e":{"kind":"REWORK","_number":30,"created":"2025-06-25 11:07:32.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/30"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-25 11:07:21.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the creation of new\ninspection rules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/e59ca38a90774a9dd86be44ae893dfcfb9c44b5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/e59ca38a90774a9dd86be44ae893dfcfb9c44b5e"}]},"branch":"refs/heads/master"},"ea3ed3a7266da8c4a03ca97c09809da5c3232745":{"kind":"NO_CHANGE","_number":31,"created":"2025-06-30 13:46:50.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/31"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-30 13:46:34.000000000","tz":60},"subject":"WIP Add Tempest tests for inspection rules in Ironic","message":"WIP Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the creation of new\ninspection rules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/ea3ed3a7266da8c4a03ca97c09809da5c3232745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/ea3ed3a7266da8c4a03ca97c09809da5c3232745"}]},"branch":"refs/heads/master"},"ae7a77df9c0849d1c9f32c096657a650ccc721ee":{"kind":"REWORK","_number":32,"created":"2025-07-02 08:57:32.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/32"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-07-02 08:56:29.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the creation of new\ninspection rules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\nSigned-off-by: abhibongale \u003cabhishekbongale@outlook.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/ae7a77df9c0849d1c9f32c096657a650ccc721ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/ae7a77df9c0849d1c9f32c096657a650ccc721ee"}]},"branch":"refs/heads/master"},"7fa87559b59df31aa351c40156eaabc67e14e062":{"kind":"REWORK","_number":33,"created":"2025-07-03 12:42:18.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/33"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-07-03 12:42:06.000000000","tz":60},"subject":"Add Tempest tests for inspection rules in Ironic","message":"Add Tempest tests for inspection rules in Ironic\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the creation of new\ninspection rules functionality within Ironic.\n\nThe tests cover the basic CRUD operations for inspection rules\nand ensure they are correctly applied during node introspection.\n\nThese tests are intended to ensure parity with existing functionality\nand provide coverage for the integrated implementation now within\nIronic.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\nSigned-off-by: abhibongale \u003cabhishekbongale@outlook.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/7fa87559b59df31aa351c40156eaabc67e14e062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/7fa87559b59df31aa351c40156eaabc67e14e062"}]},"branch":"refs/heads/master"},"fa630483afe3b469c2eddeb9aba57ddf3c807d66":{"kind":"NO_CODE_CHANGE","_number":34,"created":"2025-07-07 16:08:01.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/61/951761/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-tempest-plugin","ref":"refs/changes/61/951761/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-tempest-plugin refs/changes/61/951761/34"}}},"commit":{"parents":[{"commit":"8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3","subject":"Merge \"Adding better error messages to microversion tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/8ea34bfb77b62964e587d9ccf36c74b1d3fd12b3"}]}],"author":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-06-04 12:30:29.000000000","tz":60},"committer":{"name":"abhibongale","email":"abhishekbongale@outlook.com","date":"2025-07-07 15:48:47.000000000","tz":60},"subject":"Add Tempest API tests for inspection rules","message":"Add Tempest API tests for inspection rules\n\nAs of the 2025.1 \"Epoxy\" release, inspection rules have been migrated\nfrom the ironic-inspector project into Ironic itself.\n\nThis commit introduces Tempest tests to validate the creation, updation\nand deletion of new inspection rules functionality within Ironic.\n\nCurrent focus is on covering the important CRUD operations for\ninspection rules to ensure robustness and correctness of\nthe Inspection rules API.\n\nI guess this is the starting point, as Inspection rules evolves\nmore tests should be added to ensure the continued relevance\neffectiveness of the API test suite.\n\ncloses-bug: 2105478\n\nChange-Id: I3f6de02acee8d8c3764d3b1465b92292be3b690c\nSigned-off-by: abhibongale \u003cabhishekbongale@outlook.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/fa630483afe3b469c2eddeb9aba57ddf3c807d66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-tempest-plugin/commit/fa630483afe3b469c2eddeb9aba57ddf3c807d66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
