)]}'
{"id":"openstack%2Fironic-ui~617803","triplet_id":"openstack%2Fironic-ui~master~I0d8ad20827142d2d2cc9c453768b0356860f1255","project":"openstack/ironic-ui","branch":"master","hashtags":[],"change_id":"I0d8ad20827142d2d2cc9c453768b0356860f1255","subject":"Update various links in the documentation","status":"MERGED","created":"2018-11-13 19:36:09.000000000","updated":"2019-09-04 14:04:50.000000000","submitted":"2019-09-04 14:03:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"617803-1567605795761-233c4fa7","meta_rev_id":"a27b6a92ef937d32ad1efb46d40957ae967f754f","_number":617803,"virtual_id_number":617803,"owner":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2019-09-04 14:03:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2019-09-04 13:48:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2019-09-04 13:48:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-19 10:35:52.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-09-04 13:48:47.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-09-04 14:03:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf7b3f61d306b1d20a405e6c5eb2aaadad165e6b","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2018-11-13 19:36:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc936342d81930967ffaaf5f2100d43b4de8b166","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-13 19:48:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/617803/1/check/openstack-tox-pep8/00376a3/ : SUCCESS in 5m 06s\n- nodejs-npm-run-lint http://logs.openstack.org/03/617803/1/check/nodejs-npm-run-lint/fd0d53f/ : SUCCESS in 6m 05s\n- nodejs-npm-run-test http://logs.openstack.org/03/617803/1/check/nodejs-npm-run-test/fcef8ca/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/03/617803/1/check/openstack-tox-docs/fbd18d4/html/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f732cc63a96b93efffbb8577ac693bbe91edd08d","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2018-11-15 11:12:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f2dc35c6d56f9d4018f4b229cee0e99d744bbead","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2018-11-15 13:41:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"403a9495107306274aa816bfa520f1c9988b383b","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2018-11-15 13:42:08.000000000","message":"Patch Set 2:\n\n\u003e (2 comments)\n\n \u003e Thank you very much for your advice.","accounts_in_message":[],"_revision_number":2},{"id":"de672f70900e3b9eeb69866719db16426a1cfa89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-15 13:53:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/617803/2/check/openstack-tox-pep8/7e92d1a/ : SUCCESS in 5m 13s\n- nodejs-npm-run-lint http://logs.openstack.org/03/617803/2/check/nodejs-npm-run-lint/2853a5f/ : SUCCESS in 5m 54s\n- nodejs-npm-run-test http://logs.openstack.org/03/617803/2/check/nodejs-npm-run-test/b80b5f6/ : SUCCESS in 6m 41s\n- openstack-tox-docs http://logs.openstack.org/03/617803/2/check/openstack-tox-docs/5a421ae/html/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"1fca5835d5c253fd2ae6b07255c9ddd24e26c102","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2018-11-19 10:35:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c56b63b801b757b2e598e87d9ec2a568e9d042e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-04 12:24:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"64a94cf944a7bc8e49ebdce0a3b56b823799af57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 12:40:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/2f250dc177c5402baa52c0ce0ff67213 : SUCCESS in 5m 26s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/6c46c94534fa47e284accaf42e92752f : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf541279d2294bfb88784dca10e331f5 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/257c6b75a8d04c91ad8888f02a2a6d56 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"e315acd1209b098f7d0fd3a4ca2508423aead2e6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-09-04 13:48:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"14550c9746514d011ca22aa172d5843912f09abe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:48:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"50564ddfdedee64bb4405a2443a1edfbdaf180dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:03:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/b7b9be44800949979e2727d581bc7a2e : SUCCESS in 7m 18s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/02c98af9db014880906afb8ef18575e4 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55bcfd86267d440d90443d28bb82943b : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3dbdac1ed647a5acc30da360e78bcf : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"d57c055456fd25911b5390c59e6c601e2e395245","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:03:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a27b6a92ef937d32ad1efb46d40957ae967f754f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:04:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8653c441d044aa88328ff4fae3fb646 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ce04b57f14275a75097e78bd02254f8003a974e9","revisions":{"10c16514705d088081943c84ccb35a6fea6b0f94":{"kind":"REWORK","_number":1,"created":"2018-11-13 19:36:09.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/03/617803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/03/617803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/1"}}},"commit":{"parents":[{"commit":"59aa382b37fe7d63ec2c16d69ca3786dd1ea3d06","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/59aa382b37fe7d63ec2c16d69ca3786dd1ea3d06"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2018-11-13 19:34:20.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2018-11-13 19:35:59.000000000","tz":480},"subject":"Replacing the HTTP protocal with HTTPS in installation.rst.","message":"Replacing the HTTP protocal with HTTPS in installation.rst.\n\nChange-Id: I0d8ad20827142d2d2cc9c453768b0356860f1255\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/10c16514705d088081943c84ccb35a6fea6b0f94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/10c16514705d088081943c84ccb35a6fea6b0f94"}]},"branch":"refs/heads/master"},"48e5ed8d502075f2fe49677d90b97b277bbe328e":{"kind":"REWORK","_number":2,"created":"2018-11-15 13:41:27.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/03/617803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/03/617803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/2"}}},"commit":{"parents":[{"commit":"59aa382b37fe7d63ec2c16d69ca3786dd1ea3d06","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/59aa382b37fe7d63ec2c16d69ca3786dd1ea3d06"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2018-11-13 19:34:20.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2018-11-15 13:41:22.000000000","tz":0},"subject":"Replacing the HTTP protocal with HTTPS.","message":"Replacing the HTTP protocal with HTTPS.\n\nChange-Id: I0d8ad20827142d2d2cc9c453768b0356860f1255\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/48e5ed8d502075f2fe49677d90b97b277bbe328e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/48e5ed8d502075f2fe49677d90b97b277bbe328e"}]},"branch":"refs/heads/master"},"ce04b57f14275a75097e78bd02254f8003a974e9":{"kind":"REWORK","_number":3,"created":"2019-09-04 12:24:55.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/03/617803/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/03/617803/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/03/617803/3"}}},"commit":{"parents":[{"commit":"3f95d588f7cf38e3dda006ef7bee0525377ac872","subject":"Merge \"Follow the new PTI for document build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/3f95d588f7cf38e3dda006ef7bee0525377ac872"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2018-11-13 19:34:20.000000000","tz":480},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-04 12:24:53.000000000","tz":120},"subject":"Update various links in the documentation","message":"Update various links in the documentation\n\nCo-Authored-By: Dmitry Tantsur \u003cdtantsur@redhat.com\u003e\nChange-Id: I0d8ad20827142d2d2cc9c453768b0356860f1255\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/ce04b57f14275a75097e78bd02254f8003a974e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/ce04b57f14275a75097e78bd02254f8003a974e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
