)]}'
{"id":"openstack%2Fironic-ui~654882","triplet_id":"openstack%2Fironic-ui~master~I8a306d1b42cb25296960f88158eb8c8e21b0be08","project":"openstack/ironic-ui","branch":"master","hashtags":[],"change_id":"I8a306d1b42cb25296960f88158eb8c8e21b0be08","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 09:00:08.000000000","updated":"2019-06-07 04:35:05.000000000","submitted":"2019-04-25 09:59:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654882-1556186354490-7a5aba1f","meta_rev_id":"f3f9b697e18454f6ed25d086f99795579ecf6001","_number":654882,"virtual_id_number":654882,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-04-25 09:59:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"all":[{"value":1,"date":"2019-04-24 16:42:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-04-24 17:02:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-04-24 07:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-25 09:47:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-25 09:47:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 07:53:14.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-04-24 17:02:22.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-04-25 09:47:01.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-04-25 09:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e88b49fbf6fc81d53d12778a63debb8a32ba8b8f","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-23 09:00:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b72f0e7d8b5cfc3fc29ab0d42587eed5273ab4cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:45:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/82/654882/1/check/test-release-openstack/eed0225/ : SUCCESS in 1m 56s\n- nodejs-npm-run-lint http://logs.openstack.org/82/654882/1/check/nodejs-npm-run-lint/c93f047/ : RETRY_LIMIT in 3m 38s\n- nodejs-npm-run-test http://logs.openstack.org/82/654882/1/check/nodejs-npm-run-test/a8f11e1/ : RETRY_LIMIT in 3m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/654882/1/check/openstack-tox-lower-constraints/f0021c0/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/82/654882/1/check/openstack-tox-pep8/eeec1de/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/82/654882/1/check/openstack-tox-py27/b40b082/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/82/654882/1/check/openstack-tox-py36/fbdc890/ : SUCCESS in 3m 47s\n- openstack-tox-py37 http://logs.openstack.org/82/654882/1/check/openstack-tox-py37/f304ea9/ : SUCCESS in 4m 31s\n- openstack-tox-docs http://logs.openstack.org/82/654882/1/check/openstack-tox-docs/c8b52db/html/ : SUCCESS in 3m 26s\n- build-openstack-releasenotes http://logs.openstack.org/82/654882/1/check/build-openstack-releasenotes/8996bb8/html/ : SUCCESS in 4m 22s\n- ironic-ui-integration-tests http://logs.openstack.org/82/654882/1/check/ironic-ui-integration-tests/1418b1b/ : FAILURE in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec05a43908d045dfd5801424139fa8b3b8a51ecd","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-23 13:04:11.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fd6bbe9884cb1bf82769d6eba8d0c47bb434702","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-23 17:13:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a469e9ad0cca785f31cdb39c7f6c18096158b77c","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-23 17:14:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aac1bdd9997692ea664c72e0f606475df4b42c35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 20:31:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/82/654882/2/check/test-release-openstack/934beaf/ : SUCCESS in 1m 54s\n- nodejs-npm-run-lint http://logs.openstack.org/82/654882/2/check/nodejs-npm-run-lint/195fade/ : SUCCESS in 4m 56s\n- nodejs-npm-run-test http://logs.openstack.org/82/654882/2/check/nodejs-npm-run-test/84ddbf3/ : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/654882/2/check/openstack-tox-lower-constraints/2925ce5/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/82/654882/2/check/openstack-tox-pep8/46fff9d/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/82/654882/2/check/openstack-tox-py27/13909b5/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/82/654882/2/check/openstack-tox-py36/66ed8f6/ : SUCCESS in 3m 23s\n- openstack-tox-py37 http://logs.openstack.org/82/654882/2/check/openstack-tox-py37/1c7748c/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/82/654882/2/check/openstack-tox-docs/216e978/html/ : SUCCESS in 3m 21s\n- build-openstack-releasenotes http://logs.openstack.org/82/654882/2/check/build-openstack-releasenotes/69e550d/html/ : SUCCESS in 3m 52s\n- ironic-ui-integration-tests http://logs.openstack.org/82/654882/2/check/ironic-ui-integration-tests/c9eacc3/ : SUCCESS in 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24c2206a65816c6f6e1833b77dc25e2010bca0c8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-24 07:53:14.000000000","message":"Patch Set 2: Code-Review+2\n\nlooks good, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"7239facb72502e2d11eaa076766563d16f35dc1f","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-24 16:42:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"39ce1fcd28ca7b5d983a923a38a3fc135945c4ab","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-24 17:02:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c1ac67a59a38a922d5fcdfba9c467531141b4822","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-04-25 09:47:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a6bd11bf38e8db7ad5b421898f088d7b70a7b01e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 09:47:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b0e51a78b598c6a16b00d705a354ba78017a0ce8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 09:59:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9a7fe822db03b872dcaad46c11aea5ffd00e00ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 09:59:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/82/654882/2/gate/test-release-openstack/a0573b3/ : SUCCESS in 1m 50s\n- nodejs-npm-run-lint http://logs.openstack.org/82/654882/2/gate/nodejs-npm-run-lint/1324b82/ : SUCCESS in 5m 32s\n- nodejs-npm-run-test http://logs.openstack.org/82/654882/2/gate/nodejs-npm-run-test/d762d5b/ : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/654882/2/gate/openstack-tox-lower-constraints/859693d/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/82/654882/2/gate/openstack-tox-pep8/2d66f60/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/82/654882/2/gate/openstack-tox-py27/14e8216/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/82/654882/2/gate/openstack-tox-py36/14e2d43/ : SUCCESS in 3m 17s\n- openstack-tox-py37 http://logs.openstack.org/82/654882/2/gate/openstack-tox-py37/22e33b4/ : SUCCESS in 3m 26s\n- openstack-tox-docs http://logs.openstack.org/82/654882/2/gate/openstack-tox-docs/eb590f6/html/ : SUCCESS in 3m 06s\n- build-openstack-releasenotes http://logs.openstack.org/82/654882/2/gate/build-openstack-releasenotes/95f78e8/html/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f3f9b697e18454f6ed25d086f99795579ecf6001","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-07 04:35:05.000000000","message":"Patch Set 2:\n\nthanks","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6670f4339a76c87b55d515c81f8239c00164cb8f","revisions":{"eabb3c1a115a36f9429c72fc6fa107fcd2880349":{"kind":"REWORK","_number":1,"created":"2019-04-23 09:00:08.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/82/654882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/82/654882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/654882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/654882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/654882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/82/654882/1"}}},"commit":{"parents":[{"commit":"96543ce0a6980e8ebe9425adc0c250de81fff766","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/96543ce0a6980e8ebe9425adc0c250de81fff766"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:59:56.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 09:00:06.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I8a306d1b42cb25296960f88158eb8c8e21b0be08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/eabb3c1a115a36f9429c72fc6fa107fcd2880349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/eabb3c1a115a36f9429c72fc6fa107fcd2880349"}]},"branch":"refs/heads/master"},"6670f4339a76c87b55d515c81f8239c00164cb8f":{"kind":"REWORK","_number":2,"created":"2019-04-23 17:13:44.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/82/654882/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/82/654882/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/654882/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/654882/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/654882/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/82/654882/2"}}},"commit":{"parents":[{"commit":"96543ce0a6980e8ebe9425adc0c250de81fff766","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/96543ce0a6980e8ebe9425adc0c250de81fff766"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:59:56.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 17:13:36.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I8a306d1b42cb25296960f88158eb8c8e21b0be08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/6670f4339a76c87b55d515c81f8239c00164cb8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/6670f4339a76c87b55d515c81f8239c00164cb8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
