)]}'
{"id":"openstack%2Fironic-ui~656932","triplet_id":"openstack%2Fironic-ui~stable%2Fstein~I02de1b094eac5ea2f37588b7fd5cc4cf74c701ea","project":"openstack/ironic-ui","branch":"stable/stein","topic":"zanata/translations","hashtags":[],"change_id":"I02de1b094eac5ea2f37588b7fd5cc4cf74c701ea","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-05-03 08:55:59.000000000","updated":"2019-05-03 19:59:22.000000000","submitted":"2019-05-03 19:59:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656932-1556913562802-959ddcf6","meta_rev_id":"7d2bcbcaff5d22951898e37d6b4a11e29e7b08ea","_number":656932,"virtual_id_number":656932,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-05-03 19:59:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2019-05-03 10:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-03 19:49:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-03 19:49:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 10:39:56.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-05-03 19:49:26.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-05-03 19:59:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03fddc91b1b2957c6d070afc05e277b78f3a07e8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-03 08:55:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3d25a8dab54efdf2b6c6098286c85ebe73432ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 09:32:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/32/656932/1/check/nodejs-npm-run-lint/6aa8544/ : SUCCESS in 4m 52s\n- nodejs-npm-run-test http://logs.openstack.org/32/656932/1/check/nodejs-npm-run-test/528778e/ : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/656932/1/check/openstack-tox-lower-constraints/420b7bd/ : SUCCESS in 3m 30s\n- openstack-tox-pep8 http://logs.openstack.org/32/656932/1/check/openstack-tox-pep8/56b3de3/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/32/656932/1/check/openstack-tox-py27/aa8f1a6/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/32/656932/1/check/openstack-tox-py35/070ea77/ : SUCCESS in 3m 17s\n- openstack-tox-py36 http://logs.openstack.org/32/656932/1/check/openstack-tox-py36/55ed48f/ : SUCCESS in 3m 14s\n- openstack-tox-py37 http://logs.openstack.org/32/656932/1/check/openstack-tox-py37/19b83ce/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/32/656932/1/check/openstack-tox-docs/92a4e98/html/ : SUCCESS in 3m 20s\n- ironic-ui-integration-tests http://logs.openstack.org/32/656932/1/check/ironic-ui-integration-tests/f7b5682/ : SUCCESS in 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cd9dc9edf87331d3ccc437b63b64d655d03256b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-03 10:39:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b913e0f980bcee45651ac5b128e60c21c26e8581","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-03 19:49:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d179e9d3eb7e5f3220a4d1854f15ec71a99fdfee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 19:49:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f9fbbe76ec2c272978d1b33f4903dde07607d70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 19:59:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a42eefa9abd56ed32ca8f9e647a470279d313c18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 19:59:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/32/656932/1/gate/nodejs-npm-run-lint/ba5be81/ : SUCCESS in 4m 23s\n- nodejs-npm-run-test http://logs.openstack.org/32/656932/1/gate/nodejs-npm-run-test/1893de2/ : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/656932/1/gate/openstack-tox-lower-constraints/5049680/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/32/656932/1/gate/openstack-tox-pep8/3b1e99c/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/32/656932/1/gate/openstack-tox-py27/d89d83d/ : SUCCESS in 3m 17s\n- openstack-tox-py35 http://logs.openstack.org/32/656932/1/gate/openstack-tox-py35/e7502cd/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/32/656932/1/gate/openstack-tox-py36/ce1d79e/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/32/656932/1/gate/openstack-tox-py37/76ede9a/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/32/656932/1/gate/openstack-tox-docs/19e4931/html/ : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9bd177319a9198834cbaa7f47025090828bedad","revisions":{"d9bd177319a9198834cbaa7f47025090828bedad":{"kind":"REWORK","_number":1,"created":"2019-05-03 08:55:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/656932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/32/656932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/32/656932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/32/656932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/32/656932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/32/656932/1"}}},"commit":{"parents":[{"commit":"c4ac3f0c19584fdfd398fd635baa429191d5834e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/c4ac3f0c19584fdfd398fd635baa429191d5834e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-03 08:55:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-03 08:55:57.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I02de1b094eac5ea2f37588b7fd5cc4cf74c701ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/d9bd177319a9198834cbaa7f47025090828bedad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/d9bd177319a9198834cbaa7f47025090828bedad"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
