)]}'
{"id":"openstack%2Fironic-ui~657097","triplet_id":"openstack%2Fironic-ui~stable%2Fstein~Ie0d63651afbfff032fab78d747884877137e7d26","project":"openstack/ironic-ui","branch":"stable/stein","topic":"zanata/translations","hashtags":[],"change_id":"Ie0d63651afbfff032fab78d747884877137e7d26","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-05-04 08:52:38.000000000","updated":"2019-05-08 17:04:08.000000000","submitted":"2019-05-08 17:04:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657097-1557335048740-8ca2a984","meta_rev_id":"04a54cce61d8d1a0b499a922b68be017ac0293bb","_number":657097,"virtual_id_number":657097,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-05-08 17:04:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2019-05-06 08:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-08 16:45:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-08 16:45:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 08:43:20.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-05-08 16:45:57.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-05-08 17:04:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e66e84750c54ffb66257f96a4b608ed0b2e0fe5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-04 08:52:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91cc3c26b47c893506b95ef2ff25d9ffda7e41c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 09:32:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/97/657097/1/check/nodejs-npm-run-lint/fa9bdb0/ : SUCCESS in 4m 06s\n- nodejs-npm-run-test http://logs.openstack.org/97/657097/1/check/nodejs-npm-run-test/ca73f72/ : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/657097/1/check/openstack-tox-lower-constraints/7c990dd/ : SUCCESS in 2m 57s\n- openstack-tox-pep8 http://logs.openstack.org/97/657097/1/check/openstack-tox-pep8/a613e18/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/97/657097/1/check/openstack-tox-py27/f0bb634/ : SUCCESS in 2m 46s\n- openstack-tox-py35 http://logs.openstack.org/97/657097/1/check/openstack-tox-py35/413d777/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/97/657097/1/check/openstack-tox-py36/985ea4b/ : SUCCESS in 2m 59s\n- openstack-tox-py37 http://logs.openstack.org/97/657097/1/check/openstack-tox-py37/0e3d2c0/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/97/657097/1/check/openstack-tox-docs/67ed015/html/ : SUCCESS in 3m 03s\n- ironic-ui-integration-tests http://logs.openstack.org/97/657097/1/check/ironic-ui-integration-tests/3dbd7e2/ : SUCCESS in 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"750182bd05457ac46290fcfc7bea1618cbba1b0c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-06 08:43:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33ba4a469911f53ca755b99b626b67d04928e3d1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-08 16:45:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"584ba5ee16d42312b36650f99741cfe937035654","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 16:46:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"63bb40aacdd858b85149a91cb8fd80939eb03a00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 17:04:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"79191cc8f40be10054c6d6061f0ee8ad2f57ca54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 17:04:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/97/657097/1/gate/nodejs-npm-run-lint/d07a11d/ : SUCCESS in 4m 25s\n- nodejs-npm-run-test http://logs.openstack.org/97/657097/1/gate/nodejs-npm-run-test/cd5acf3/ : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/657097/1/gate/openstack-tox-lower-constraints/4de1fd3/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/97/657097/1/gate/openstack-tox-pep8/f79cb74/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/97/657097/1/gate/openstack-tox-py27/5cb18a6/ : SUCCESS in 3m 23s\n- openstack-tox-py35 http://logs.openstack.org/97/657097/1/gate/openstack-tox-py35/2b96b12/ : SUCCESS in 3m 39s\n- openstack-tox-py36 http://logs.openstack.org/97/657097/1/gate/openstack-tox-py36/eee21a6/ : SUCCESS in 4m 01s\n- openstack-tox-py37 http://logs.openstack.org/97/657097/1/gate/openstack-tox-py37/3d5c60c/ : SUCCESS in 3m 44s\n- openstack-tox-docs http://logs.openstack.org/97/657097/1/gate/openstack-tox-docs/0eb8a79/html/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"859560fe8722edbd61a79d3137814295fad43b61","revisions":{"859560fe8722edbd61a79d3137814295fad43b61":{"kind":"REWORK","_number":1,"created":"2019-05-04 08:52:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/657097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/97/657097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/97/657097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/97/657097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/97/657097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/97/657097/1"}}},"commit":{"parents":[{"commit":"d9bd177319a9198834cbaa7f47025090828bedad","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/d9bd177319a9198834cbaa7f47025090828bedad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-04 08:52:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-04 08:52:36.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ie0d63651afbfff032fab78d747884877137e7d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/859560fe8722edbd61a79d3137814295fad43b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/859560fe8722edbd61a79d3137814295fad43b61"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
