)]}'
{"id":"openstack%2Fironic-ui~657358","triplet_id":"openstack%2Fironic-ui~master~I44c43eec521c564ce14678481a7759cef6106270","project":"openstack/ironic-ui","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I44c43eec521c564ce14678481a7759cef6106270","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-05-06 09:06:29.000000000","updated":"2019-05-08 16:57:49.000000000","submitted":"2019-05-08 16:57:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657358-1557334669246-ef73585b","meta_rev_id":"c59e822d0f2891464f8efb5983ab0ec5b83a2877","_number":657358,"virtual_id_number":657358,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-05-08 16:57:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2019-05-06 09:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-08 16:45:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-08 16:45:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 09:20:31.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-05-08 16:45:52.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-05-08 16:57:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2699bdf4377d5ffdb88ee48381dd9b20a361e16b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-06 09:06:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed2a2b74c9f52a2d34d7839a3716330288155868","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-06 09:20:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b1fa7ba233647ab4821b66e84e033e6f6d4426b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 09:51:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/58/657358/1/check/nodejs-npm-run-lint/5306435/ : SUCCESS in 5m 33s\n- nodejs-npm-run-test http://logs.openstack.org/58/657358/1/check/nodejs-npm-run-test/d288fea/ : SUCCESS in 7m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/657358/1/check/openstack-tox-lower-constraints/7b54a90/ : SUCCESS in 3m 15s\n- openstack-tox-pep8 http://logs.openstack.org/58/657358/1/check/openstack-tox-pep8/81ed329/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/58/657358/1/check/openstack-tox-py27/c7fb805/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/58/657358/1/check/openstack-tox-py36/0003ce9/ : SUCCESS in 3m 08s\n- openstack-tox-py37 http://logs.openstack.org/58/657358/1/check/openstack-tox-py37/d9a5946/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/58/657358/1/check/openstack-tox-docs/f24bc4c/html/ : SUCCESS in 3m 17s\n- ironic-ui-integration-tests http://logs.openstack.org/58/657358/1/check/ironic-ui-integration-tests/4f5c510/ : SUCCESS in 40m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"761840fcc6eb41b3729ce21f7c1d496f7dfc8138","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-08 16:45:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"064b15a9e4e0e6444c35234d0dddae2f78f7db86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 16:46:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91737eeccabcfa7798de104f73c69de026e29881","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 16:57:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4eef05531cbec70ff3507d344fc7f6c55e162ee6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 16:57:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/58/657358/1/gate/nodejs-npm-run-lint/7719204/ : SUCCESS in 4m 43s\n- nodejs-npm-run-test http://logs.openstack.org/58/657358/1/gate/nodejs-npm-run-test/f0572ad/ : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/657358/1/gate/openstack-tox-lower-constraints/501f008/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/58/657358/1/gate/openstack-tox-pep8/24639ea/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/58/657358/1/gate/openstack-tox-py27/fc1a5ac/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/58/657358/1/gate/openstack-tox-py36/087f105/ : SUCCESS in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/58/657358/1/gate/openstack-tox-py37/fa344c2/ : SUCCESS in 4m 36s\n- openstack-tox-docs http://logs.openstack.org/58/657358/1/gate/openstack-tox-docs/c65e7ac/html/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11ca90e75da518a96cabfc1a96e03d7b813a43a9","revisions":{"11ca90e75da518a96cabfc1a96e03d7b813a43a9":{"kind":"REWORK","_number":1,"created":"2019-05-06 09:06:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/58/657358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/58/657358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/58/657358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/58/657358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/58/657358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/58/657358/1"}}},"commit":{"parents":[{"commit":"6670f4339a76c87b55d515c81f8239c00164cb8f","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/6670f4339a76c87b55d515c81f8239c00164cb8f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-06 09:06:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-06 09:06:28.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I44c43eec521c564ce14678481a7759cef6106270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/11ca90e75da518a96cabfc1a96e03d7b813a43a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/11ca90e75da518a96cabfc1a96e03d7b813a43a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
