)]}'
{"id":"openstack%2Fironic-ui~660595","triplet_id":"openstack%2Fironic-ui~master~I1f802d9a114b93b47afc8b6ebb6bd92148945a50","project":"openstack/ironic-ui","branch":"master","topic":"Bug#ironicclient_version","hashtags":[],"change_id":"I1f802d9a114b93b47afc8b6ebb6bd92148945a50","subject":"Exclude broken ironicclient versions 2.5.2 and 2.7.1","status":"MERGED","created":"2019-05-22 08:11:36.000000000","updated":"2019-05-23 08:53:34.000000000","submitted":"2019-05-23 08:53:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660595-1558601614510-d92ef69a","meta_rev_id":"020b64e9129e6c5a08c035aa22a30babda0ba348","_number":660595,"virtual_id_number":660595,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-05-23 08:53:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2019-05-22 08:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-23 08:41:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-23 08:41:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 08:17:21.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-05-23 08:41:00.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-05-23 08:53:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88edc3d38138ad9127fabd7cb5b343b6a56b5c55","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-05-22 08:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00f1959473dc6e45a961981aeb41d830b6a8ed9b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-22 08:17:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a3a7fa0f9b26cbff845e33beba84d10e737ab7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 08:49:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/95/660595/1/check/requirements-check/137e391/ : SUCCESS in 1m 50s\n- nodejs-npm-run-lint http://logs.openstack.org/95/660595/1/check/nodejs-npm-run-lint/4cd5ce6/ : SUCCESS in 4m 38s\n- nodejs-npm-run-test http://logs.openstack.org/95/660595/1/check/nodejs-npm-run-test/03b815c/ : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/660595/1/check/openstack-tox-lower-constraints/31dd367/ : SUCCESS in 3m 01s\n- openstack-tox-pep8 http://logs.openstack.org/95/660595/1/check/openstack-tox-pep8/ee4ca32/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/95/660595/1/check/openstack-tox-py27/5d2bf83/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/95/660595/1/check/openstack-tox-py36/03c2250/ : SUCCESS in 2m 58s\n- openstack-tox-py37 http://logs.openstack.org/95/660595/1/check/openstack-tox-py37/16400d5/ : SUCCESS in 3m 22s\n- openstack-tox-docs http://logs.openstack.org/95/660595/1/check/openstack-tox-docs/97fcfe8/html/ : SUCCESS in 3m 18s\n- ironic-ui-integration-tests http://logs.openstack.org/95/660595/1/check/ironic-ui-integration-tests/d53e505/ : SUCCESS in 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91fcb3da4f870705cdae0713fee0af28583e1033","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-23 08:41:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a8db343ffda45cc19a2d489bbdc296de800b118","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 08:41:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5883dc8a4b44eba66826985ca39338d12189b14d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 08:53:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"04107298b7117fe12185318691a7f78432417620","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 08:53:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/95/660595/1/gate/requirements-check/ac67442/ : SUCCESS in 2m 08s\n- nodejs-npm-run-lint http://logs.openstack.org/95/660595/1/gate/nodejs-npm-run-lint/9c5f870/ : SUCCESS in 4m 50s\n- nodejs-npm-run-test http://logs.openstack.org/95/660595/1/gate/nodejs-npm-run-test/a9e4576/ : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/660595/1/gate/openstack-tox-lower-constraints/10349b6/ : SUCCESS in 2m 56s\n- openstack-tox-pep8 http://logs.openstack.org/95/660595/1/gate/openstack-tox-pep8/7a42c98/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/95/660595/1/gate/openstack-tox-py27/864d433/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/95/660595/1/gate/openstack-tox-py36/b03a670/ : SUCCESS in 3m 21s\n- openstack-tox-py37 http://logs.openstack.org/95/660595/1/gate/openstack-tox-py37/f82a8e5/ : SUCCESS in 3m 38s\n- openstack-tox-docs http://logs.openstack.org/95/660595/1/gate/openstack-tox-docs/8e09118/html/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62851c74af3af2d804d0ea6ee78e1cdb781301dc","revisions":{"62851c74af3af2d804d0ea6ee78e1cdb781301dc":{"kind":"REWORK","_number":1,"created":"2019-05-22 08:11:36.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/95/660595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/95/660595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/95/660595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/95/660595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/95/660595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/95/660595/1"}}},"commit":{"parents":[{"commit":"72320f449579199b89bfba2842a3b5f86e6c640b","subject":"Update sphinx dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/72320f449579199b89bfba2842a3b5f86e6c640b"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-22 07:46:52.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-22 07:46:59.000000000","tz":480},"subject":"Exclude broken ironicclient versions 2.5.2 and 2.7.1","message":"Exclude broken ironicclient versions 2.5.2 and 2.7.1\n\ndepend on https://review.opendev.org/#/c/659612/\n\nChange-Id: I1f802d9a114b93b47afc8b6ebb6bd92148945a50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/62851c74af3af2d804d0ea6ee78e1cdb781301dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/62851c74af3af2d804d0ea6ee78e1cdb781301dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
