)]}'
{"id":"openstack%2Fironic-ui~667282","triplet_id":"openstack%2Fironic-ui~master~I6aba9d03ad3cadba7bbd41ea32091a751889db17","project":"openstack/ironic-ui","branch":"master","topic":"docs","hashtags":[],"change_id":"I6aba9d03ad3cadba7bbd41ea32091a751889db17","subject":"Follow the new PTI for document build","status":"MERGED","created":"2019-06-25 08:05:26.000000000","updated":"2019-08-23 18:06:16.000000000","submitted":"2019-08-23 18:04:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667282-1566583487347-7e4db9f1","meta_rev_id":"4cc3be769bb50b9613a7f61146b28cdce828d54b","_number":667282,"virtual_id_number":667282,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-08-23 18:04:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"date":"2019-08-22 15:42:06.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2019-08-15 12:09:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-23 03:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-23 17:44:36.000000000","_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-08-23 17:44:36.000000000","_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 06:26:48.000000000","updated_by":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"reviewer":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"state":"REVIEWER"},{"updated":"2019-06-27 06:26:48.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2019-06-27 06:26:48.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"},{"updated":"2019-06-27 06:26:48.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-06-27 06:26:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-27 06:26:48.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"},{"updated":"2019-07-01 07:16:43.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-15 12:09:19.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2019-08-22 15:42:06.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-08-23 03:56:08.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-08-23 17:44:36.000000000","updated_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"reviewer":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"state":"REVIEWER"},{"updated":"2019-08-23 18:04:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5588dbcdee544033b6303ed336e69cd49b346399","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-25 08:05:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89b6fd57b50ff6fcbc17d2062d90a4d31d5e2c87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 08:48:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/82/667282/1/check/test-release-openstack/d38646a/ : SUCCESS in 1m 43s\n- requirements-check http://logs.openstack.org/82/667282/1/check/requirements-check/ac970f3/ : SUCCESS in 2m 02s\n- nodejs-npm-run-lint http://logs.openstack.org/82/667282/1/check/nodejs-npm-run-lint/7a28ae5/ : SUCCESS in 5m 15s\n- nodejs-npm-run-test http://logs.openstack.org/82/667282/1/check/nodejs-npm-run-test/ffa0d1a/ : SUCCESS in 8m 05s\n- nodejs10-npm-run-lint http://logs.openstack.org/82/667282/1/check/nodejs10-npm-run-lint/52b4543/ : SUCCESS in 6m 31s\n- nodejs10-npm-run-test http://logs.openstack.org/82/667282/1/check/nodejs10-npm-run-test/a0496b6/ : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/667282/1/check/openstack-tox-lower-constraints/f803b22/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/82/667282/1/check/openstack-tox-pep8/28dece6/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/82/667282/1/check/openstack-tox-py27/4a65662/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/82/667282/1/check/openstack-tox-py36/8292b57/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/82/667282/1/check/openstack-tox-py37/6fe8b73/ : SUCCESS in 3m 55s\n- openstack-tox-docs http://logs.openstack.org/82/667282/1/check/openstack-tox-docs/be4a5f4/ : POST_FAILURE in 3m 43s\n- build-openstack-releasenotes http://logs.openstack.org/82/667282/1/check/build-openstack-releasenotes/645de55/html/ : SUCCESS in 3m 34s\n- ironic-ui-integration-tests http://logs.openstack.org/82/667282/1/check/ironic-ui-integration-tests/f4fc998/ : SUCCESS in 39m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6dc2ad059cb0a18645df7bcaebe65b689c62516","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-25 09:05:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"14e10ea6d5f009fe984d1b10e6ab151722626f93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 11:59:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/82/667282/2/check/test-release-openstack/b4c280d/ : SUCCESS in 4m 28s\n- requirements-check http://logs.openstack.org/82/667282/2/check/requirements-check/c9f1553/ : SUCCESS in 4m 18s\n- nodejs-npm-run-lint http://logs.openstack.org/82/667282/2/check/nodejs-npm-run-lint/211c134/ : SUCCESS in 6m 29s\n- nodejs-npm-run-test http://logs.openstack.org/82/667282/2/check/nodejs-npm-run-test/bfd43a9/ : SUCCESS in 13m 51s\n- nodejs10-npm-run-lint http://logs.openstack.org/82/667282/2/check/nodejs10-npm-run-lint/e0dc202/ : SUCCESS in 7m 52s\n- nodejs10-npm-run-test http://logs.openstack.org/82/667282/2/check/nodejs10-npm-run-test/0b9565d/ : SUCCESS in 8m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/667282/2/check/openstack-tox-lower-constraints/bd881d2/ : SUCCESS in 5m 35s\n- openstack-tox-pep8 http://logs.openstack.org/82/667282/2/check/openstack-tox-pep8/ce964a7/ : SUCCESS in 13m 34s\n- openstack-tox-py27 http://logs.openstack.org/82/667282/2/check/openstack-tox-py27/a97d1a9/ : SUCCESS in 5m 34s\n- openstack-tox-py36 http://logs.openstack.org/82/667282/2/check/openstack-tox-py36/c5d57f9/ : SUCCESS in 13m 52s\n- openstack-tox-py37 http://logs.openstack.org/82/667282/2/check/openstack-tox-py37/0aaea81/ : SUCCESS in 6m 23s\n- openstack-tox-docs http://logs.openstack.org/82/667282/2/check/openstack-tox-docs/fff2933/ : FAILURE in 14m 18s\n- build-openstack-releasenotes http://logs.openstack.org/82/667282/2/check/build-openstack-releasenotes/512df31/html/ : SUCCESS in 5m 34s\n- ironic-ui-integration-tests http://logs.openstack.org/82/667282/2/check/ironic-ui-integration-tests/1a607cd/ : SUCCESS in 41m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8584ae26c634b026ed78f84cea89be4d5b3539a1","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-26 01:34:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"83a4421004a582951af3849e04ed6c61a923c95e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 02:15:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/82/667282/3/check/test-release-openstack/5571ac7/ : SUCCESS in 1m 44s\n- requirements-check http://logs.openstack.org/82/667282/3/check/requirements-check/a7b9b29/ : SUCCESS in 2m 14s\n- nodejs-npm-run-lint http://logs.openstack.org/82/667282/3/check/nodejs-npm-run-lint/5d0e6b2/ : SUCCESS in 5m 06s\n- nodejs-npm-run-test http://logs.openstack.org/82/667282/3/check/nodejs-npm-run-test/adcae58/ : SUCCESS in 5m 53s\n- nodejs10-npm-run-lint http://logs.openstack.org/82/667282/3/check/nodejs10-npm-run-lint/7bc0d86/ : SUCCESS in 5m 01s\n- nodejs10-npm-run-test http://logs.openstack.org/82/667282/3/check/nodejs10-npm-run-test/9897822/ : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/667282/3/check/openstack-tox-lower-constraints/0aa4557/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/82/667282/3/check/openstack-tox-pep8/9088a15/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/82/667282/3/check/openstack-tox-py27/41fb98d/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/82/667282/3/check/openstack-tox-py36/b824d64/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/82/667282/3/check/openstack-tox-py37/a028356/ : SUCCESS in 3m 55s\n- openstack-tox-docs http://logs.openstack.org/82/667282/3/check/openstack-tox-docs/6c1faa0/ : FAILURE in 3m 41s\n- build-openstack-releasenotes http://logs.openstack.org/82/667282/3/check/build-openstack-releasenotes/191a44a/html/ : SUCCESS in 4m 37s\n- ironic-ui-integration-tests http://logs.openstack.org/82/667282/3/check/ironic-ui-integration-tests/0fbbbaf/ : SUCCESS in 35m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3d7a635452fbea5d7543d6cccabbc52c4a0c7460","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-26 05:46:01.000000000","message":"Topic changed from I6aba9d03ad3cadba7bbd41ea32091a751889db17 to doc","accounts_in_message":[],"_revision_number":3},{"id":"7e49bdcdf373b526bbe4054c1172818007083b55","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-26 05:50:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f13d2b034614604e446fc8c621abdd5b5e21832e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 06:37:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/82/667282/3/check/test-release-openstack/32f4869/ : SUCCESS in 1m 58s\n- requirements-check http://logs.openstack.org/82/667282/3/check/requirements-check/96345df/ : SUCCESS in 2m 04s\n- nodejs-npm-run-lint http://logs.openstack.org/82/667282/3/check/nodejs-npm-run-lint/b6e0d20/ : SUCCESS in 5m 50s\n- nodejs-npm-run-test http://logs.openstack.org/82/667282/3/check/nodejs-npm-run-test/fe70626/ : SUCCESS in 7m 43s\n- nodejs10-npm-run-lint http://logs.openstack.org/82/667282/3/check/nodejs10-npm-run-lint/937eeee/ : SUCCESS in 5m 51s\n- nodejs10-npm-run-test http://logs.openstack.org/82/667282/3/check/nodejs10-npm-run-test/feb512f/ : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/667282/3/check/openstack-tox-lower-constraints/6a56033/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/82/667282/3/check/openstack-tox-pep8/da2c307/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/82/667282/3/check/openstack-tox-py27/3e85f8f/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/82/667282/3/check/openstack-tox-py36/9abf661/ : SUCCESS in 4m 12s\n- openstack-tox-py37 http://logs.openstack.org/82/667282/3/check/openstack-tox-py37/f43b306/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/82/667282/3/check/openstack-tox-docs/517ba11/ : FAILURE in 4m 02s\n- build-openstack-releasenotes http://logs.openstack.org/82/667282/3/check/build-openstack-releasenotes/1369490/html/ : SUCCESS in 4m 46s\n- ironic-ui-integration-tests http://logs.openstack.org/82/667282/3/check/ironic-ui-integration-tests/e9b8cb7/ : SUCCESS in 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"74c788e18ec9396a81b6cee3339cc484ded7908c","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-26 06:57:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"30f8bb16994889ec8197abf0af8522c60fb715bb","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-26 07:29:23.000000000","message":"Topic changed from I6aba9d03ad3cadba7bbd41ea32091a751889db17 to docs","accounts_in_message":[],"_revision_number":4},{"id":"e69f9f6b0bed4fc45f1f4d1bffbbcd1d9261d467","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 07:58:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/82/667282/4/check/test-release-openstack/580aa71/ : SUCCESS in 2m 06s\n- requirements-check http://logs.openstack.org/82/667282/4/check/requirements-check/ef93711/ : SUCCESS in 1m 55s\n- nodejs-npm-run-lint http://logs.openstack.org/82/667282/4/check/nodejs-npm-run-lint/154e4a6/ : SUCCESS in 5m 20s\n- nodejs-npm-run-test http://logs.openstack.org/82/667282/4/check/nodejs-npm-run-test/5fcf3e9/ : SUCCESS in 6m 46s\n- nodejs10-npm-run-lint http://logs.openstack.org/82/667282/4/check/nodejs10-npm-run-lint/e3e83e5/ : SUCCESS in 4m 31s\n- nodejs10-npm-run-test http://logs.openstack.org/82/667282/4/check/nodejs10-npm-run-test/b4e5c45/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/667282/4/check/openstack-tox-lower-constraints/a353ab4/ : SUCCESS in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/82/667282/4/check/openstack-tox-pep8/041618a/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/82/667282/4/check/openstack-tox-py27/dec7ea2/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/82/667282/4/check/openstack-tox-py36/6bcf162/ : SUCCESS in 3m 18s\n- openstack-tox-py37 http://logs.openstack.org/82/667282/4/check/openstack-tox-py37/bd21bc0/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/82/667282/4/check/openstack-tox-docs/dbbd2e4/ : FAILURE in 3m 12s\n- build-openstack-releasenotes http://logs.openstack.org/82/667282/4/check/build-openstack-releasenotes/4d84a05/html/ : SUCCESS in 4m 16s\n- ironic-ui-integration-tests http://logs.openstack.org/82/667282/4/check/ironic-ui-integration-tests/d48a548/ : SUCCESS in 38m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f22dc31b56dc60d1ae91655a7a2e5d18c76e7040","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-27 01:38:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"54afee25e20cd70c0f085e0899d2e3c3a0d9c4f8","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-27 01:38:58.000000000","message":"Topic changed from I6aba9d03ad3cadba7bbd41ea32091a751889db17 to doc","accounts_in_message":[],"_revision_number":5},{"id":"f8ae2e754c4591d5718f27cb029cc291f21f027d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 02:25:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/82/667282/5/check/test-release-openstack/bd95418/ : SUCCESS in 2m 06s\n- requirements-check http://logs.openstack.org/82/667282/5/check/requirements-check/70a94fd/ : SUCCESS in 3m 12s\n- nodejs-npm-run-lint http://logs.openstack.org/82/667282/5/check/nodejs-npm-run-lint/67e7497/ : SUCCESS in 5m 22s\n- nodejs-npm-run-test http://logs.openstack.org/82/667282/5/check/nodejs-npm-run-test/c7e42d8/ : SUCCESS in 6m 11s\n- nodejs10-npm-run-lint http://logs.openstack.org/82/667282/5/check/nodejs10-npm-run-lint/56b789c/ : SUCCESS in 4m 46s\n- nodejs10-npm-run-test http://logs.openstack.org/82/667282/5/check/nodejs10-npm-run-test/1b04406/ : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/667282/5/check/openstack-tox-lower-constraints/e7f5b8d/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/82/667282/5/check/openstack-tox-pep8/8437c30/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/82/667282/5/check/openstack-tox-py27/1dd0be5/ : SUCCESS in 3m 16s\n- openstack-tox-py36 http://logs.openstack.org/82/667282/5/check/openstack-tox-py36/6dbe745/ : SUCCESS in 3m 35s\n- openstack-tox-py37 http://logs.openstack.org/82/667282/5/check/openstack-tox-py37/ecb09af/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/82/667282/5/check/openstack-tox-docs/616ba24/ : FAILURE in 4m 19s\n- build-openstack-releasenotes http://logs.openstack.org/82/667282/5/check/build-openstack-releasenotes/4f422ea/html/ : SUCCESS in 4m 16s\n- ironic-ui-integration-tests http://logs.openstack.org/82/667282/5/check/ironic-ui-integration-tests/fc3c990/ : SUCCESS in 39m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aaa6cb94132c1bbaa16f07b00b52f0909c35d813","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-27 02:28:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"34de3e012707844653836633baf3cc2e223674c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:14:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/82/667282/6/check/test-release-openstack/e8e014b/ : SUCCESS in 2m 04s\n- requirements-check http://logs.openstack.org/82/667282/6/check/requirements-check/5cce96d/ : SUCCESS in 2m 17s\n- nodejs-npm-run-lint http://logs.openstack.org/82/667282/6/check/nodejs-npm-run-lint/26a5571/ : SUCCESS in 5m 36s\n- nodejs-npm-run-test http://logs.openstack.org/82/667282/6/check/nodejs-npm-run-test/83382f0/ : FAILURE in 6m 55s\n- nodejs10-npm-run-lint http://logs.openstack.org/82/667282/6/check/nodejs10-npm-run-lint/4ea67a4/ : SUCCESS in 6m 41s\n- nodejs10-npm-run-test http://logs.openstack.org/82/667282/6/check/nodejs10-npm-run-test/fb5f464/ : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/667282/6/check/openstack-tox-lower-constraints/659a5ec/ : SUCCESS in 6m 34s\n- openstack-tox-pep8 http://logs.openstack.org/82/667282/6/check/openstack-tox-pep8/d7f5c34/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/82/667282/6/check/openstack-tox-py27/1418b7b/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/82/667282/6/check/openstack-tox-py36/7decf47/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/82/667282/6/check/openstack-tox-py37/69c6f93/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/82/667282/6/check/openstack-tox-docs/daa2433/html/ : SUCCESS in 4m 15s\n- build-openstack-releasenotes http://logs.openstack.org/82/667282/6/check/build-openstack-releasenotes/3dc6328/html/ : SUCCESS in 4m 00s\n- ironic-ui-integration-tests http://logs.openstack.org/82/667282/6/check/ironic-ui-integration-tests/053ae33/ : SUCCESS in 38m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c8f0de04b7054c3a2c640f72157594b0009529d8","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-27 03:20:18.000000000","message":"Topic changed from I6aba9d03ad3cadba7bbd41ea32091a751889db17 to docs","accounts_in_message":[],"_revision_number":6},{"id":"eb7945d286fc5e2a1c0c13eefa899e894f395a45","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-27 03:20:28.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"de8bb02779668f4ee15704bd41fd0349226d76fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 04:08:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/82/667282/6/check/test-release-openstack/3d6793a/ : SUCCESS in 2m 22s\n- requirements-check http://logs.openstack.org/82/667282/6/check/requirements-check/afb021b/ : SUCCESS in 3m 13s\n- nodejs-npm-run-lint http://logs.openstack.org/82/667282/6/check/nodejs-npm-run-lint/3a2afc3/ : SUCCESS in 6m 02s\n- nodejs-npm-run-test http://logs.openstack.org/82/667282/6/check/nodejs-npm-run-test/7e28a55/ : SUCCESS in 6m 23s\n- nodejs10-npm-run-lint http://logs.openstack.org/82/667282/6/check/nodejs10-npm-run-lint/521b5c8/ : SUCCESS in 4m 48s\n- nodejs10-npm-run-test http://logs.openstack.org/82/667282/6/check/nodejs10-npm-run-test/ba4cb2e/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/667282/6/check/openstack-tox-lower-constraints/7894c99/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/82/667282/6/check/openstack-tox-pep8/a74aba9/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/82/667282/6/check/openstack-tox-py27/73d4ea3/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/82/667282/6/check/openstack-tox-py36/ef34faf/ : SUCCESS in 3m 31s\n- openstack-tox-py37 http://logs.openstack.org/82/667282/6/check/openstack-tox-py37/f35affd/ : SUCCESS in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/82/667282/6/check/openstack-tox-docs/07dbd74/html/ : SUCCESS in 3m 35s\n- build-openstack-releasenotes http://logs.openstack.org/82/667282/6/check/build-openstack-releasenotes/32568f2/html/ : SUCCESS in 5m 34s\n- ironic-ui-integration-tests http://logs.openstack.org/82/667282/6/check/ironic-ui-integration-tests/499c126/ : SUCCESS in 36m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"aca8983d984f9600446b74ea2fee9b500dd189f6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-28 15:23:25.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"744be9e452dd61c2a9e4576d8c74b9adc96940e0","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-01 02:59:26.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"01146af9133802ab4b152a51c32b0160af6187dd","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-01 02:59:54.000000000","message":"Topic changed from I6aba9d03ad3cadba7bbd41ea32091a751889db17 to docs","accounts_in_message":[],"_revision_number":7},{"id":"03bff9e5630db49e1b4cf74faf57e559ba2c482e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 03:40:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/82/667282/7/check/test-release-openstack/802d3e7/ : SUCCESS in 1m 57s\n- requirements-check http://logs.openstack.org/82/667282/7/check/requirements-check/8956969/ : SUCCESS in 1m 59s\n- nodejs-npm-run-lint http://logs.openstack.org/82/667282/7/check/nodejs-npm-run-lint/dfcc1c7/ : SUCCESS in 5m 25s\n- nodejs-npm-run-test http://logs.openstack.org/82/667282/7/check/nodejs-npm-run-test/6f75491/ : SUCCESS in 6m 05s\n- nodejs10-npm-run-lint http://logs.openstack.org/82/667282/7/check/nodejs10-npm-run-lint/d3ad293/ : SUCCESS in 4m 11s\n- nodejs10-npm-run-test http://logs.openstack.org/82/667282/7/check/nodejs10-npm-run-test/1d8b8fc/ : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/667282/7/check/openstack-tox-lower-constraints/3bf2605/ : SUCCESS in 3m 11s\n- openstack-tox-pep8 http://logs.openstack.org/82/667282/7/check/openstack-tox-pep8/dd65f07/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/82/667282/7/check/openstack-tox-py27/6d6fb5d/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/82/667282/7/check/openstack-tox-py36/f9a20b4/ : SUCCESS in 3m 16s\n- openstack-tox-py37 http://logs.openstack.org/82/667282/7/check/openstack-tox-py37/068290b/ : SUCCESS in 3m 23s\n- openstack-tox-docs http://logs.openstack.org/82/667282/7/check/openstack-tox-docs/50e0e4d/html/ : SUCCESS in 3m 08s\n- build-openstack-releasenotes http://logs.openstack.org/82/667282/7/check/build-openstack-releasenotes/7a0c06c/html/ : SUCCESS in 3m 52s\n- ironic-ui-integration-tests http://logs.openstack.org/82/667282/7/check/ironic-ui-integration-tests/09f6710/ : SUCCESS in 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2e10b562227f48b617d6b9ef7b766f954f27288d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-01 07:16:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d1b76eccf459f5d68451478c9f7613e6b523f0c8","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-02 06:15:30.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"5640ea0e79eef1823a30c2f306c49bb15e107423","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-02 06:15:48.000000000","message":"Topic changed from I6aba9d03ad3cadba7bbd41ea32091a751889db17 to docs","accounts_in_message":[],"_revision_number":8},{"id":"0f08b9ee7163a3b4978d9f0eda5e04e8997b5bef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 07:11:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/82/667282/8/check/test-release-openstack/6e26e6f/ : SUCCESS in 2m 10s\n- requirements-check http://logs.openstack.org/82/667282/8/check/requirements-check/7dff3c9/ : SUCCESS in 2m 23s\n- nodejs-npm-run-lint http://logs.openstack.org/82/667282/8/check/nodejs-npm-run-lint/7e008a5/ : SUCCESS in 5m 23s\n- nodejs-npm-run-test http://logs.openstack.org/82/667282/8/check/nodejs-npm-run-test/c99c561/ : SUCCESS in 6m 47s\n- nodejs10-npm-run-lint http://logs.openstack.org/82/667282/8/check/nodejs10-npm-run-lint/1d2dd67/ : SUCCESS in 5m 42s\n- nodejs10-npm-run-test http://logs.openstack.org/82/667282/8/check/nodejs10-npm-run-test/baf0f6f/ : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/667282/8/check/openstack-tox-lower-constraints/722be51/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/82/667282/8/check/openstack-tox-pep8/1dacf85/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/82/667282/8/check/openstack-tox-py27/285ff27/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/82/667282/8/check/openstack-tox-py36/9582bf7/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/82/667282/8/check/openstack-tox-py37/30f05aa/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/82/667282/8/check/openstack-tox-docs/4ca818d/html/ : SUCCESS in 3m 30s\n- build-openstack-releasenotes http://logs.openstack.org/82/667282/8/check/build-openstack-releasenotes/20eb0cb/html/ : SUCCESS in 3m 49s\n- ironic-ui-integration-tests http://logs.openstack.org/82/667282/8/check/ironic-ui-integration-tests/7d58680/ : SUCCESS in 36m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8178eacab08fba3e2d06f5f36fc03761571fe81c","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-08-15 12:09:19.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"e64aa580a348b7a0a209084a39db7156d1db2452","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-08-22 15:42:06.000000000","message":"Patch Set 8:\n\nrecheck stale results","accounts_in_message":[],"_revision_number":8},{"id":"fbbedd40be99d029167a2359ab3a0fe3d9cf3d1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 16:36:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/590ac8d5afe74b5b99e2f47d96550ced : SUCCESS in 2m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/df6429cbd12a4b70b8e987e2573be6fa : SUCCESS in 3m 47s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/0ebe7d957616485d9e44b50d027ba929 : SUCCESS in 7m 15s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/bbebaa027f8b49e285f11a498845043e : SUCCESS in 7m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6753d8753ac74b32af6301e56d1ad16f : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/572956a68c4846dcb374ee6e5c625553 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72219b46d32f452ab9e53ef399c4eb42 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94bece8862034202b66cebeb0785e208 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e31b4069be2e42c7ad8e58c54e878463 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/628f193e01214ac89845401d6461ea52 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e59572686a4e437b89e7343616507f70 : SUCCESS in 9m 14s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/5a954dbf9bad4369a054fef957085262 : SUCCESS in 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0575c6228d74a34564bc4c4da5fb914a42036a6c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-08-23 03:56:08.000000000","message":"Patch Set 8: Code-Review+2\n\nlooks good, thanks","accounts_in_message":[],"_revision_number":8},{"id":"270f1ebd6bf3849e7451798a43bc019398e3ddf2","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2019-08-23 17:44:36.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(2 comments)\n\nThanks!\n\nI looked at the generated docs and they seem fine, so we can iterate over this if I brought up any interesting questions :)","accounts_in_message":[],"_revision_number":8},{"id":"a48c2ce27cac981e18b49337cf40dad038372380","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 17:44:48.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"545063a1e13a4d2f8e08e30221ec70e9f2b2d611","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 18:04:47.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4d6ca56da8004d1f8261e39756731189 : SUCCESS in 3m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/56356c857a904e4883c560936fe8b829 : SUCCESS in 2m 36s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/fbdc1f9a713e4cb29cc51bc566e31009 : SUCCESS in 7m 54s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/37357d29ffab41139e5bd16cc507b794 : SUCCESS in 8m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b62f926b7719435a86930d229862cc1e : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4427e6f1fb3644dbb40ef8008120f660 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75933a0106204ee782f38cb346eb5841 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cb98fcf227f4007992d1af7c8220fe0 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c46fc8108ef4f638c06721b674d622c : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f340425f4e184d57bcc6d12de8f90ad1 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/073389b534a94301a259071af213e915 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":8},{"id":"44e29c4656cb174915173804fcc9aff7a3893842","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 18:04:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"4cc3be769bb50b9613a7f61146b28cdce828d54b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 18:06:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b411ab858de47b384da90036351d420 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"b68684a3ef6f75f65246066c28d1af879844741d","revisions":{"83f38a6254dbd4ac985915816dd38dd524739362":{"kind":"REWORK","_number":1,"created":"2019-06-25 08:05:26.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/82/667282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/82/667282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/1"}}},"commit":{"parents":[{"commit":"62e4f97c484c3e67115a67373b08d1d3bc3db0a5","subject":"Add nodejs10 jobs on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/62e4f97c484c3e67115a67373b08d1d3bc3db0a5"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 07:45:40.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 07:45:53.000000000","tz":480},"subject":"Follow the new PTI for document build","message":"Follow the new PTI for document build\n\nThe detail is described in the following links:\nhttps://governance.openstack.org/tc/reference/project-testing-interface.html\nhttp://lists.openstack.org/pipermail/openstack-dev/2017-December/125710.html\n\ndoc8 command is dropped from docs and releasenotes tox envs.\nWe run doc8 in pep8 tox env, so this hurts nothing.\n\nChange-Id: I6aba9d03ad3cadba7bbd41ea32091a751889db17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/83f38a6254dbd4ac985915816dd38dd524739362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/83f38a6254dbd4ac985915816dd38dd524739362"}]},"branch":"refs/heads/master"},"bfdc79cce72f118972584c7d95dedc1c48095903":{"kind":"REWORK","_number":2,"created":"2019-06-25 09:05:08.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/82/667282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/82/667282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/2"}}},"commit":{"parents":[{"commit":"62e4f97c484c3e67115a67373b08d1d3bc3db0a5","subject":"Add nodejs10 jobs on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/62e4f97c484c3e67115a67373b08d1d3bc3db0a5"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 07:45:40.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 08:45:48.000000000","tz":480},"subject":"Follow the new PTI for document build","message":"Follow the new PTI for document build\n\nThe detail is described in the following links:\nhttps://governance.openstack.org/tc/reference/project-testing-interface.html\nhttp://lists.openstack.org/pipermail/openstack-dev/2017-December/125710.html\n\ndoc8 command is dropped from docs and releasenotes tox envs.\nWe run doc8 in pep8 tox env, so this hurts nothing.\n\nChange-Id: I6aba9d03ad3cadba7bbd41ea32091a751889db17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/bfdc79cce72f118972584c7d95dedc1c48095903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/bfdc79cce72f118972584c7d95dedc1c48095903"}]},"branch":"refs/heads/master"},"1306601cdd344d650a58a22ce2f80cacd15096b4":{"kind":"REWORK","_number":3,"created":"2019-06-26 01:34:52.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/82/667282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/82/667282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/3"}}},"commit":{"parents":[{"commit":"62e4f97c484c3e67115a67373b08d1d3bc3db0a5","subject":"Add nodejs10 jobs on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/62e4f97c484c3e67115a67373b08d1d3bc3db0a5"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 07:45:40.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-26 01:15:34.000000000","tz":480},"subject":"Follow the new PTI for document build","message":"Follow the new PTI for document build\n\nThe detail is described in the following links:\nhttps://governance.openstack.org/tc/reference/project-testing-interface.html\nhttp://lists.openstack.org/pipermail/openstack-dev/2017-December/125710.html\n\ndoc8 command is dropped from docs and releasenotes tox envs.\nWe run doc8 in pep8 tox env, so this hurts nothing.\n\nChange-Id: I6aba9d03ad3cadba7bbd41ea32091a751889db17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/1306601cdd344d650a58a22ce2f80cacd15096b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/1306601cdd344d650a58a22ce2f80cacd15096b4"}]},"branch":"refs/heads/master"},"7c68271c029962e39b54cc596a4f211bb76b6878":{"kind":"REWORK","_number":4,"created":"2019-06-26 06:57:18.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/82/667282/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/82/667282/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/4"}}},"commit":{"parents":[{"commit":"62e4f97c484c3e67115a67373b08d1d3bc3db0a5","subject":"Add nodejs10 jobs on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/62e4f97c484c3e67115a67373b08d1d3bc3db0a5"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 07:45:40.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-26 06:38:02.000000000","tz":480},"subject":"Follow the new PTI for document build","message":"Follow the new PTI for document build\n\nThe detail is described in the following links:\nhttps://governance.openstack.org/tc/reference/project-testing-interface.html\nhttp://lists.openstack.org/pipermail/openstack-dev/2017-December/125710.html\n\ndoc8 command is dropped from docs and releasenotes tox envs.\nWe run doc8 in pep8 tox env, so this hurts nothing.\n\nChange-Id: I6aba9d03ad3cadba7bbd41ea32091a751889db17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/7c68271c029962e39b54cc596a4f211bb76b6878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/7c68271c029962e39b54cc596a4f211bb76b6878"}]},"branch":"refs/heads/master"},"17d093e045251c297dae9cdf4f9bbb75db2f1e5f":{"kind":"REWORK","_number":5,"created":"2019-06-27 01:38:32.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/82/667282/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/82/667282/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/5"}}},"commit":{"parents":[{"commit":"62e4f97c484c3e67115a67373b08d1d3bc3db0a5","subject":"Add nodejs10 jobs on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/62e4f97c484c3e67115a67373b08d1d3bc3db0a5"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 07:45:40.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-27 01:19:10.000000000","tz":480},"subject":"Follow the new PTI for document build","message":"Follow the new PTI for document build\n\nThe detail is described in the following links:\nhttps://governance.openstack.org/tc/reference/project-testing-interface.html\nhttp://lists.openstack.org/pipermail/openstack-dev/2017-December/125710.html\n\ndoc8 command is dropped from docs and releasenotes tox envs.\nWe run doc8 in pep8 tox env, so this hurts nothing.\n\nChange-Id: I6aba9d03ad3cadba7bbd41ea32091a751889db17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/17d093e045251c297dae9cdf4f9bbb75db2f1e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/17d093e045251c297dae9cdf4f9bbb75db2f1e5f"}]},"branch":"refs/heads/master"},"77da1609a8b06324b6bb457d4c6da9a15f71818f":{"kind":"REWORK","_number":6,"created":"2019-06-27 02:28:42.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/82/667282/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/82/667282/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/6"}}},"commit":{"parents":[{"commit":"62e4f97c484c3e67115a67373b08d1d3bc3db0a5","subject":"Add nodejs10 jobs on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/62e4f97c484c3e67115a67373b08d1d3bc3db0a5"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 07:45:40.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-27 02:09:22.000000000","tz":480},"subject":"Follow the new PTI for document build","message":"Follow the new PTI for document build\n\nThe detail is described in the following links:\nhttps://governance.openstack.org/tc/reference/project-testing-interface.html\nhttp://lists.openstack.org/pipermail/openstack-dev/2017-December/125710.html\n\ndoc8 command is dropped from docs and releasenotes tox envs.\nWe run doc8 in pep8 tox env, so this hurts nothing.\n\nChange-Id: I6aba9d03ad3cadba7bbd41ea32091a751889db17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/77da1609a8b06324b6bb457d4c6da9a15f71818f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/77da1609a8b06324b6bb457d4c6da9a15f71818f"}]},"branch":"refs/heads/master"},"09be4b76eb17c793817ab58a56f7a0a6e7404ad2":{"kind":"REWORK","_number":7,"created":"2019-07-01 02:59:26.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/82/667282/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/82/667282/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/7"}}},"commit":{"parents":[{"commit":"62e4f97c484c3e67115a67373b08d1d3bc3db0a5","subject":"Add nodejs10 jobs on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/62e4f97c484c3e67115a67373b08d1d3bc3db0a5"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 07:45:40.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-01 02:59:16.000000000","tz":480},"subject":"Follow the new PTI for document build","message":"Follow the new PTI for document build\n\nThe detail is described in the following links:\nhttps://governance.openstack.org/tc/reference/project-testing-interface.html\nhttp://lists.openstack.org/pipermail/openstack-dev/2017-December/125710.html\n\ndoc8 command is dropped from docs and releasenotes tox envs.\nWe run doc8 in pep8 tox env, so this hurts nothing.\n\nChange-Id: I6aba9d03ad3cadba7bbd41ea32091a751889db17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/09be4b76eb17c793817ab58a56f7a0a6e7404ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/09be4b76eb17c793817ab58a56f7a0a6e7404ad2"}]},"branch":"refs/heads/master"},"b68684a3ef6f75f65246066c28d1af879844741d":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-07-02 06:15:30.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/82/667282/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/82/667282/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/82/667282/8"}}},"commit":{"parents":[{"commit":"62e4f97c484c3e67115a67373b08d1d3bc3db0a5","subject":"Add nodejs10 jobs on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/62e4f97c484c3e67115a67373b08d1d3bc3db0a5"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-25 07:45:40.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-02 06:15:19.000000000","tz":480},"subject":"Follow the new PTI for document build","message":"Follow the new PTI for document build\n\nThe detail is described in the following links:\nhttps://governance.openstack.org/tc/reference/project-testing-interface.html\nhttp://lists.openstack.org/pipermail/openstack-dev/2017-December/125710.html\n\nChange-Id: I6aba9d03ad3cadba7bbd41ea32091a751889db17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/b68684a3ef6f75f65246066c28d1af879844741d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/b68684a3ef6f75f65246066c28d1af879844741d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
