)]}'
{"id":"openstack%2Fironic-ui~672767","triplet_id":"openstack%2Fironic-ui~master~Ic11c2ad91331eef7e59e9e9995d273d0af025ca6","project":"openstack/ironic-ui","branch":"master","topic":"retire-api-site","hashtags":[],"change_id":"Ic11c2ad91331eef7e59e9e9995d273d0af025ca6","subject":"Update api-ref location","status":"MERGED","created":"2019-07-25 16:04:31.000000000","updated":"2019-07-26 09:22:44.000000000","submitted":"2019-07-26 09:22:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672767-1564132964927-622e802a","meta_rev_id":"b1cc694a1d7420f35fda98db6cd35538b277cbec","_number":672767,"virtual_id_number":672767,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":2,"date":"2019-07-26 09:22:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":2,"date":"2019-07-26 09:11:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-25 17:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":1,"date":"2019-07-26 09:11:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-25 17:08:04.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-07-26 09:11:13.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"},{"updated":"2019-07-26 09:22:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3809ecbabfb7f0e3f24d9f0e4a330178d06984b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-25 16:04:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc418aeeac29a178856446b75c0beb3772bcd91b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 17:00:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/67/672767/1/check/nodejs10-npm-run-lint/bea2e54/ : SUCCESS in 6m 35s\n- nodejs10-npm-run-test http://logs.openstack.org/67/672767/1/check/nodejs10-npm-run-test/0b33cad/ : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/672767/1/check/openstack-tox-lower-constraints/82665af/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/67/672767/1/check/openstack-tox-pep8/ef60385/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/67/672767/1/check/openstack-tox-py27/d7019f0/ : SUCCESS in 5m 30s\n- openstack-tox-py36 http://logs.openstack.org/67/672767/1/check/openstack-tox-py36/ab867a0/ : SUCCESS in 5m 52s\n- openstack-tox-py37 http://logs.openstack.org/67/672767/1/check/openstack-tox-py37/f89b119/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/67/672767/1/check/openstack-tox-docs/23d2b50/html/ : SUCCESS in 4m 34s\n- ironic-ui-integration-tests http://logs.openstack.org/67/672767/1/check/ironic-ui-integration-tests/a53a54c/ : SUCCESS in 48m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"822e008f2f312a773517fcde75205894bdbfc65a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-25 17:08:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"67f6955b2d0ad5e79832b83cfb4a45b92266a2e1","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-07-26 09:11:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"537d374ab2754e9938c79a1e869b582f220c9f66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 09:11:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f52cef0c606a5cc1794acc00718b718bf50d4b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 09:22:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3014ab70754dea5564cc390f4143da07cdd53199","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 09:22:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/67/672767/1/gate/nodejs10-npm-run-lint/9fffd56/ : SUCCESS in 5m 27s\n- nodejs10-npm-run-test http://logs.openstack.org/67/672767/1/gate/nodejs10-npm-run-test/b4b9856/ : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/672767/1/gate/openstack-tox-lower-constraints/0f7b9cb/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/67/672767/1/gate/openstack-tox-pep8/65f77df/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/67/672767/1/gate/openstack-tox-py27/34281ed/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/67/672767/1/gate/openstack-tox-py36/b7da006/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/67/672767/1/gate/openstack-tox-py37/dfa60f4/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/67/672767/1/gate/openstack-tox-docs/3ef318c/html/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"292cc2ecebd499534da8f12c2f2320c60963dace","revisions":{"292cc2ecebd499534da8f12c2f2320c60963dace":{"kind":"REWORK","_number":1,"created":"2019-07-25 16:04:31.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/67/672767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/67/672767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/67/672767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/67/672767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/67/672767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/67/672767/1"}}},"commit":{"parents":[{"commit":"9e80cc80e930cdbeec6cfe29d47acaf9616f4fe3","subject":"Drop nodejs4 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/9e80cc80e930cdbeec6cfe29d47acaf9616f4fe3"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-25 16:04:21.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-25 16:04:21.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: Ic11c2ad91331eef7e59e9e9995d273d0af025ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/292cc2ecebd499534da8f12c2f2320c60963dace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/292cc2ecebd499534da8f12c2f2320c60963dace"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
