)]}'
{"id":"openstack%2Fironic-ui~695801","triplet_id":"openstack%2Fironic-ui~master~Iaedcd8e73871583e4dc16860a81eb7351ffae0a2","project":"openstack/ironic-ui","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Iaedcd8e73871583e4dc16860a81eb7351ffae0a2","subject":"Drop python 2.7 support and testing","status":"MERGED","created":"2019-11-23 14:00:10.000000000","updated":"2019-11-28 10:13:36.000000000","submitted":"2019-11-28 10:10:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695801-1574935854450-c4d0231e","meta_rev_id":"b8dd5a832a151e4a353a8aa3233ce3054428a788","_number":695801,"virtual_id_number":695801,"owner":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-28 10:10:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2019-11-27 18:38:13.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2019-11-27 19:27:14.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-11-28 10:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":1,"date":"2019-11-27 18:40:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2019-11-28 10:02:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 18:38:13.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2019-11-27 18:40:03.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2019-11-27 19:27:14.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-11-28 10:02:03.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-11-28 10:10:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9bed48d2b215a487d6e9817fa1684cc18c37433","author":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"date":"2019-11-23 14:00:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d281412a88962e224ee05bcb6640fde26b3d590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 14:39:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0de84e308202470d89b6572f28b34a25 : SUCCESS in 1m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7efa8b878cdb4900999cda9f925e2b5b : SUCCESS in 1m 54s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/8f49f69cc23a410daf41ebcb05d57efc : SUCCESS in 4m 12s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/31a22ae65d094350ad1fdb7ae6494fb1 : FAILURE in 5m 26s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/e615b6d976484d85a8eb2bb362186176 : SUCCESS in 3m 42s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/cbbb5aecbe9b409cb3eae63be5a03ac9 : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c07389053fbe4cb3ba1eaefdd465a2d9 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9d621b4a5d249a9af1a42b70b50166a : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ac8adcd032344a58d1a9f26974701ff : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6e1bf8fe1f44eb6801088ddefb02057 : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d792ee1f7088404ba844b9146b137bc8 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/509e142049594d179b1614bef2c0e9ea : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba20709e21634ee99204e61c43859770 : SUCCESS in 5m 03s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/1307b2994cbc4cfea8a1558231ce711b : SUCCESS in 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4387997707a5c82a03cacd727a2532cc0d08ee44","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-25 11:10:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0176ef1b3456aa04b079b3bad101b856ca178112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 11:55:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/82a63517cf844727a54a490572a30d04 : SUCCESS in 3m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d8c4256a2bfd478880fa497f8fbd5c20 : SUCCESS in 2m 59s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/bec4185f06f44ca2b4d54425970b59bd : SUCCESS in 6m 23s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/85188fdb07bb402e89a3bfd2f894adba : FAILURE in 5m 26s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/a82340f76e174e178c99d88aaf40714d : SUCCESS in 4m 04s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/e25ff23f498f4d5d94fb18abf8cfc871 : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29492c8bdeb745d4be0e46f508c8867f : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6aad29cd1f8432da2025c39eef80002 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d14c1fe5ad0e4488814e2f887145670c : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9ed1080e8b547fb874f228c626b2484 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b6d58aba1d743089115ec3ebd346884 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52950b203f884aa8aad542427bf0117c : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4401a69c162d41c5acd0339251cf0523 : SUCCESS in 5m 13s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/def81a889e4c44578923fbe7d26fbf16 : SUCCESS in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3314fab16c3fed021bd1b8f7941c1bbc13b283f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-25 13:47:33.000000000","message":"Patch Set 1: Code-Review-1\n\nit seems the nodejs10-npm jobs are based on Python 2.7 and they don\u0027t work well with py3 as basepython; there\u0027s no equivalent yet for Python 3, so we\u0027ll have to wait for that to be available in horizon\nalso can you please remove openstack-python-jobs from .zuul.yaml ?","accounts_in_message":[],"_revision_number":1},{"id":"4e34d4d8f7fdd43d4f0181606973efae8215c4fa","author":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"date":"2019-11-27 07:01:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4620b912310930c06cd852314e419d60aa531e41","author":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"date":"2019-11-27 07:08:42.000000000","message":"Patch Set 2:\n\n\u003e it seems the nodejs10-npm jobs are based on Python 2.7 and they\n \u003e don\u0027t work well with py3 as basepython; there\u0027s no equivalent yet\n \u003e for Python 3, so we\u0027ll have to wait for that to be available in\n \u003e horizon\n \u003e also can you please remove openstack-python-jobs from .zuul.yaml ?\n\nI removed the openstack-python-jobs. Let\u0027s wait for nodejs* test to run to python3. Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"53bc436cb0af7f7e0bc20fcebb65157725e7efad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 08:14:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4f3c872e53e64640961ebb5af585086f : SUCCESS in 3m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7c74ac7f377e403cb1fb09b63ae443cc : SUCCESS in 2m 18s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/34593d9490af4131b4012b9156533cf4 : SUCCESS in 5m 43s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/d08c866ff5a5488eb32abc10042f6054 : FAILURE in 5m 58s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/6f9a79d8002448dfbe947490772f9620 : SUCCESS in 4m 38s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/f7df802b99a447b0a2052b6dbcde751f : SUCCESS in 4m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/068fbd4eb02042b3be45ba06f49d1eed : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c12b1597d64e96a89ab2a62a4a5e8e : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b91e3f51c48b449aaaa17ccc45830e33 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7dafb834bd3a4bcdaac49257d26b5703 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37c8264c4be414687d5eb18360a686b : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/153b90289a1f4d0189c3b0eb653cd5e4 : SUCCESS in 5m 33s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/47b14eb005cf4411a0b5d7a8c37988a1 : SUCCESS in 1h 09m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"164080c13a45f2079accc78a64188d6055c1b0a7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-27 10:17:16.000000000","message":"Patch Set 2: Workflow-1\n\n\u003e \u003e it seems the nodejs10-npm jobs are based on Python 2.7 and they\n \u003e \u003e don\u0027t work well with py3 as basepython; there\u0027s no equivalent yet\n \u003e \u003e for Python 3, so we\u0027ll have to wait for that to be available in\n \u003e \u003e horizon\n \u003e \u003e also can you please remove openstack-python-jobs from .zuul.yaml\n \u003e ?\n \u003e \n \u003e I removed the openstack-python-jobs. Let\u0027s wait for nodejs* test to\n \u003e run to python3. Thanks!\n\nThanks Madhuri, I\u0027m setting this as WIP until there\u0027s a proper nodejs10-npm Python 3 compatible test","accounts_in_message":[],"_revision_number":2},{"id":"52273b82767962a8f8a8f3f1c37b9530f077c579","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-11-27 14:39:05.000000000","message":"Patch Set 2:\n\nzun-ui also has the same job but they dropped Py2.7 O.o https://review.opendev.org/#/c/692008/4 maybe we should talk with people on openstack-horizon?","accounts_in_message":[],"_revision_number":2},{"id":"2273068dbe6591549ef16710707fca1af75db4f2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-11-27 17:05:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"217c0588bcfdbae0816ff2ba7a3e05be77a327c5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-27 17:12:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f82defff40b9cf2aaa1c953f55d1a4ca3ec47515","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-11-27 17:27:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e032519396eeda4a93fc1cd845d0d1dd39f138f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 17:44:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2df33a62187a453c96e71699622cd5ed : SUCCESS in 3m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b0b8c5e8f19845da8edcfcb718500236 : SUCCESS in 2m 44s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/c42e414acc6d49748d62b4702b8fffc0 : SUCCESS in 5m 25s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/758e04f8d6134f2d8f28c56e367391be : SUCCESS in 8m 17s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/fdea67f0f5684ae3aec7ea5e0e42de30 : SUCCESS in 5m 46s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/33795843cb744c9e9cc26a9bad55dad5 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4679ac47f8204320b9ef71cc0f40a948 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71cae13a188b48c688655fcf43522ec1 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37d155b590f94dc681d801a2b2b53e05 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63e9697b62444ecb83e0b60aac04c289 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9063fd5a8a3a4ee8b12e8e794ab379f7 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d50f187e993e48af848d4cb26a045fc6 : SUCCESS in 5m 50s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/878e23212e7c43aa8b677fa81e047017 : SUCCESS in 35m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4ac8c5ab517c0fdcaefd85360b112d9fee56c84f","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-11-27 17:49:57.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ff4821df67a699c1cb5d5a4cd84f67f8695ccee3","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-11-27 17:50:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e5806661c8dd5a871c9c9a693e90a67cb486f80c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:36:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/93a521bcd362487bb70871f615c264cb : SUCCESS in 2m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/45b29cc0c9b84eb7a58c08912c8835cd : SUCCESS in 2m 24s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/dcbe52d8a96b4537a40537ecdd550a72 : SUCCESS in 4m 53s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/486df4fd49a34a818459754047570410 : SUCCESS in 6m 24s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/f5264bae8d284a0ab2d15a82d81bce07 : SUCCESS in 5m 02s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/2d500cce358e4e378deb61dd93492b61 : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33a877df1e7f451da74716990597ac79 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1831cff4a1ac41b48c8443f0cf373790 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/236a992edb7b4ef7b5cb6e2a588000c4 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/599d1ea91deb40dc911f3cacff9a822e : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ee155ba0e8f460ab3634e550e21e68c : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33da45260af04e64a30c1021eae09ef1 : SUCCESS in 5m 07s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/13f2798ba7f34b4485ce067a362b48c3 : SUCCESS in 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c24ac7c180c584ac6cfbef29887177ffa0ca5024","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-11-27 18:38:13.000000000","message":"Patch Set 4: Code-Review+2\n\nLgtm.","accounts_in_message":[],"_revision_number":4},{"id":"9ad2c7c33488872638aeea53290490eb70cd50eb","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-11-27 18:40:03.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"73f72ab64cc11cffc76259e79ff2e64726d70d39","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-11-27 19:27:14.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM. Let\u0027s wait for Iroinc team to approve this","accounts_in_message":[],"_revision_number":4},{"id":"21c3eca2e292b731bf1c230d4b2be0f73738fb91","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-28 10:02:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks all, much appreciated the help on this :)","accounts_in_message":[],"_revision_number":4},{"id":"20f5873c783317e5e6e23d56d537d739e6f3102c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 10:02:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0560e179304f3b7cff42e782afa090f57ea9064f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 10:10:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"1f6deb3659bf590138713b3eae6a974707853081","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 10:10:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e525b0527a54429e927f0280fe3a816d : SUCCESS in 2m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b84d32f328764d42bfd30dfa6851823d : SUCCESS in 2m 44s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/70d9f341d6f5487b8d19f9a1c01a2f66 : SUCCESS in 4m 53s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/3fa12363ff84495bbff7020e7f9035af : SUCCESS in 6m 47s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/5523775682414005990d5c56d9dce595 : SUCCESS in 4m 39s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/a40fe598b2294f84bae7bf6fd00c08bd : SUCCESS in 3m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d40abe26780849b1b2e03df068d0b79b : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1c3d16dba1443e978be444c9bece65 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/660001bb6e6e4de98be157aa5a1c2256 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c5367fc61634f5e981f8c53d4a001f6 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad96b8c516c043298bbe528d526d8230 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f656c2c817f41bc8fa43628c5c3cdb4 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":4},{"id":"7f1c46fca86d583e4b7f02423b2af1e33ea41c54","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 10:13:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df94dd095d4f4c0dba21cd96b5f97733 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bf92bfc5e2044588854be10be15886d : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"318ed8bc9d5a97991e878db765242089a5ef2787","revisions":{"6fada834a89a59c3fd096b3ec8ecc724df0cab9c":{"kind":"REWORK","_number":1,"created":"2019-11-23 14:00:10.000000000","uploader":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"ref":"refs/changes/01/695801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/01/695801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/1"}}},"commit":{"parents":[{"commit":"416f424027b9109e3d4a7408d781f66324f52234","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/416f424027b9109e3d4a7408d781f66324f52234"}]}],"author":{"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","date":"2019-11-23 13:30:30.000000000","tz":330},"committer":{"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","date":"2019-11-23 13:31:07.000000000","tz":330},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nironic-ui is ready with python 3 and ok to drop\nthe python 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: Iaedcd8e73871583e4dc16860a81eb7351ffae0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/6fada834a89a59c3fd096b3ec8ecc724df0cab9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/6fada834a89a59c3fd096b3ec8ecc724df0cab9c"}]},"branch":"refs/heads/master"},"7dc652f37412e626b7eae39f7c45e29e3061cc35":{"kind":"REWORK","_number":2,"created":"2019-11-27 07:01:20.000000000","uploader":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"ref":"refs/changes/01/695801/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/01/695801/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/2"}}},"commit":{"parents":[{"commit":"416f424027b9109e3d4a7408d781f66324f52234","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/416f424027b9109e3d4a7408d781f66324f52234"}]}],"author":{"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","date":"2019-11-23 13:30:30.000000000","tz":330},"committer":{"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","date":"2019-11-27 06:32:07.000000000","tz":330},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nironic-ui is ready with python 3 and ok to drop\nthe python 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: Iaedcd8e73871583e4dc16860a81eb7351ffae0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/7dc652f37412e626b7eae39f7c45e29e3061cc35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/7dc652f37412e626b7eae39f7c45e29e3061cc35"}]},"branch":"refs/heads/master"},"b4a31effc91c3ed1aa9f756c897e0ab4c609eb41":{"kind":"REWORK","_number":3,"created":"2019-11-27 17:05:53.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/01/695801/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/01/695801/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/3"}}},"commit":{"parents":[{"commit":"416f424027b9109e3d4a7408d781f66324f52234","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/416f424027b9109e3d4a7408d781f66324f52234"}]}],"author":{"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","date":"2019-11-23 13:30:30.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-11-27 17:05:31.000000000","tz":0},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nironic-ui is ready with python 3 and ok to drop\nthe python 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: Iaedcd8e73871583e4dc16860a81eb7351ffae0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/b4a31effc91c3ed1aa9f756c897e0ab4c609eb41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/b4a31effc91c3ed1aa9f756c897e0ab4c609eb41"}]},"branch":"refs/heads/master"},"318ed8bc9d5a97991e878db765242089a5ef2787":{"kind":"REWORK","_number":4,"created":"2019-11-27 17:49:57.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/01/695801/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/01/695801/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/01/695801/4"}}},"commit":{"parents":[{"commit":"416f424027b9109e3d4a7408d781f66324f52234","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/416f424027b9109e3d4a7408d781f66324f52234"}]}],"author":{"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","date":"2019-11-23 13:30:30.000000000","tz":330},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2019-11-27 17:49:52.000000000","tz":0},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nironic-ui is ready with python 3 and ok to drop\nthe python 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: Iaedcd8e73871583e4dc16860a81eb7351ffae0a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/318ed8bc9d5a97991e878db765242089a5ef2787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/318ed8bc9d5a97991e878db765242089a5ef2787"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
