)]}'
{"id":"openstack%2Fironic-ui~700628","triplet_id":"openstack%2Fironic-ui~master~I6af7b37f310a8c76155c9686a78608cde0adccd8","project":"openstack/ironic-ui","branch":"master","topic":"babel-config","hashtags":[],"change_id":"I6af7b37f310a8c76155c9686a78608cde0adccd8","subject":"translation: drop babel extractor definitions","status":"MERGED","created":"2019-12-26 17:08:03.000000000","updated":"2020-01-02 12:25:06.000000000","submitted":"2020-01-02 12:23:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700628-1577967820535-71032318","meta_rev_id":"600c9b57aa55c7b70287190884e5ab478b14619a","_number":700628,"virtual_id_number":700628,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-02 12:23:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2019-12-27 19:49:53.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-01-02 09:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-02 10:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-01-02 09:53:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-27 19:49:53.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-01-02 09:53:12.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-01-02 10:37:09.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-01-02 12:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c3345644066cd4d1986981b22cdae5d5795af0c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-26 17:08:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1244ac0c4ae146234d708d8ca881aec128019fb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 17:52:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/e7e64cc487ed4924ae794bb02e0a2886 : SUCCESS in 5m 19s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/dfce53f984d04bebb7e344c20d2622f5 : SUCCESS in 6m 31s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/ab55a5b045a448bcbe7a900a140edafb : SUCCESS in 4m 43s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/a915772012b1470b8773a9b677ec644d : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/521d53116e3f44f7bf36825ef31ac382 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9974ad08d67a43479f4137c4a6615e29 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0164d5995da48209a48922bbe433082 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcab9d027b504597aafe51e76e31b04a : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa9c9967b34641cf99f4353ae68f83d4 : SUCCESS in 5m 59s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/75ca8417de95415d8d664d0664a52eb2 : SUCCESS in 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e19a3e4ca3b50251dc0f3fecbb796e8b11dfa5d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-27 19:49:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bcbf3af8b0849ddbf959f2ba5c2b03fbe9349528","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-02 09:53:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2040dbb682a549a30e2860252ad29edf502fdee6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 09:53:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"632fe1ea5dbfbefc5b050dd7a499fdd2b9b59084","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 10:28:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/2d8ed65edd7c470fa1ba097b5c7cbd8e : SUCCESS in 5m 07s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/28df4fe64c66418aa0b59d4e6cf2de67 : SUCCESS in 7m 16s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/eafe0b3b7daa4acca800c72577f6090a : FAILURE in 4m 23s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/d6de734d76a64393b5cd1eb40afdd9d8 : SUCCESS in 3m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1952494049a84a2f931c60a295c15605 : FAILURE in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cee09578f1d04e66aa1752202e56e597 : FAILURE in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a127afbf8d4140fc844c59c2252914c8 : FAILURE in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a73bd8c83874540b33246c9f4c5cd51 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a830ae8ceae0403dafddb1719cd2f9e4 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"939e2f2d55a20aa468f5cacd2acb7e7774270020","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-01-02 10:37:09.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"39ef14acf16e991fc9a0aa10fb1379bad654db80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 11:24:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/979405b154094bb3967d1b6dbc8ab406 : SUCCESS in 4m 47s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/503f35034ab2412faaf350e8fcb04d7e : SUCCESS in 5m 36s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/7528df5761e44ba89a14bd4d88cd719d : SUCCESS in 4m 31s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/8330fbc3e00e4bdea227cb85e16c8eda : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c8f15e0fd744657ae20bed6548ac15a : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c330e5a6800244f0b2292cb0992b55a5 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85c65060e9ec41a7bb6ac06a59ea4874 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd6f72d14c79408484589533f432b633 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb58e79de9b4bdbac1a70bd6b066ba1 : SUCCESS in 7m 01s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/451712ec3e2e492f843a68fafb159ea0 : SUCCESS in 41m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"161c80ec4e8af1566117eaef05a17220064e2a1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 11:24:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2bf3fa9c18c778ea8861909ff8b9a70cd8ef4e81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 12:23:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fd057b3004814c620bec696c5e778e9329fdd1b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 12:23:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/e595622d732742248565a97b2bac2dc2 : SUCCESS in 5m 06s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/8d266c19a1d44e69969118c5531906ad : SUCCESS in 6m 08s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/ccf4063104ac400ca143cc7bce404ff8 : SUCCESS in 4m 38s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/759339c71fbd420cb092052224526fac : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/083c55605ffd4d93b620fc566918b5b8 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0492ca7dd6346e6b8e935a0c30bf6e0 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c59cf9129ef4f4fb8de40458351e99c : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea7ec27f55274c51962c8a640fdb70c6 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc0fbc8e19ac4f9d9f42e229313579f3 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ab7dccd29792f25bb57c6527c0cc4a1a6202242d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 12:25:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582d3d13b2ff4057b4951ef5c8ed1fe2 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"412c3c87e91df08241f4a7526a31a83714f6e534","revisions":{"412c3c87e91df08241f4a7526a31a83714f6e534":{"kind":"REWORK","_number":1,"created":"2019-12-26 17:08:03.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/28/700628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/28/700628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/28/700628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/28/700628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/28/700628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/28/700628/1"}}},"commit":{"parents":[{"commit":"70a293d31ac4fb7f465b59e7a3c539c5629a205f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/70a293d31ac4fb7f465b59e7a3c539c5629a205f"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-26 17:03:39.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-26 17:03:39.000000000","tz":540},"subject":"translation: drop babel extractor definitions","message":"translation: drop babel extractor definitions\n\nbabel extractors are now registered via python entry points,\nso there is no need to declare babel extractors in babel configs.\n\nThis change is important to make translation work in Django 2.2.\ndjango-babel does not work with Django 2.2 and looks unmaintained\nfor over two years. The horizon team is thinking to switch the extractor\nto enmerkar (a fork of django-babel) to make extraction of translation\nstring work again near future. It is important to drop the extractor\ndefinition to make the transition smooth.\n\nChange-Id: I6af7b37f310a8c76155c9686a78608cde0adccd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/412c3c87e91df08241f4a7526a31a83714f6e534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/412c3c87e91df08241f4a7526a31a83714f6e534"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
