)]}'
{"id":"openstack%2Fironic-ui~700845","triplet_id":"openstack%2Fironic-ui~master~I29cd34d17a00eb1b576f46d6010203603b5c8165","project":"openstack/ironic-ui","branch":"master","topic":"drop-django111-support","hashtags":[],"change_id":"I29cd34d17a00eb1b576f46d6010203603b5c8165","subject":"Drop Django 1.11 support","status":"MERGED","created":"2020-01-01 20:35:59.000000000","updated":"2020-01-06 06:59:03.000000000","submitted":"2020-01-06 06:56:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700845-1578293803926-d3a5649f","meta_rev_id":"c4746806f8a62cb2794b36ff2b72df8cd220b38d","_number":700845,"virtual_id_number":700845,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-02 19:06:33.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 06:56:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-01-03 09:23:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":2,"date":"2020-01-03 09:16:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-06 06:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-06 06:43:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-03 09:16:10.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-01-03 09:23:00.000000000","updated_by":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"reviewer":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"state":"REVIEWER"},{"updated":"2020-01-06 06:43:54.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"},{"updated":"2020-01-06 06:56:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7464fad1106470123100288d0d11df89d4deb92","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-01 20:35:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0026007c77736114c3538f419afcf0801dcd8e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-01 21:10:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/aa9e7747276a49b5877a8f8b989ad73b : SUCCESS in 1m 55s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/f607a57d79934da1bec45047386da5d3 : RETRY_LIMIT in 3m 51s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/175915969d224c05bc5d6bd48f616891 : RETRY_LIMIT in 7m 58s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/7ebdf18b048845fbb627e470c3686ad3 : RETRY_LIMIT in 2m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a84b210d38184f989670fe268e8957c7 : FAILURE in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6706c631e07a4ee59a6ac24e7c80dd43 : FAILURE in 2m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb1c6e18be3c400ab70283c62267f7da : FAILURE in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4156851a2862489a95ceca942f305ece : FAILURE in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfd5662254cb48ff80ec3f3842052695 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1087e828cbd849398b0954dfc431e743 : SUCCESS in 4m 29s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/21b3cea5625e4a5fa69d9ea807df51c8 : FAILURE in 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"975f1ae4e46409fa2d257789c02174d99c278a68","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-02 09:52:32.000000000","message":"Patch Set 1: Code-Review-1\n\nhi! thanks for the patch\nif we need horizon as dependency for installation, you\u0027ll have to remove it from test-requirements","accounts_in_message":[],"_revision_number":1},{"id":"1f6633e257bb43714c38ce3f7a1e6744484d1198","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-02 19:05:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44050bdc72903b2f9ce00de1585fdebcda911873","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-02 19:06:33.000000000","message":"Patch Set 2:\n\n\u003e hi! thanks for the patch\n \u003e if we need horizon as dependency for installation, you\u0027ll have to\n \u003e remove it from test-requirements\n\nDone.\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"c7232f0384d0a59276ecf925c53ff27527d7d72c","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-02 19:27:43.000000000","message":"Topic set to drop-django111-support","accounts_in_message":[],"_revision_number":2},{"id":"820cf0d03495680c0001b78c7533de3cc688eeaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 19:55:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/246eed787f9f40d295f5c09b4c1938b9 : SUCCESS in 2m 10s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/2e6bebbc967241de9c83a2e7fa3f72b5 : SUCCESS in 5m 13s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/4edbe588e1d242cda366b34f7eacca41 : SUCCESS in 6m 01s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/fa4a0221e2754933925e7c9d9ae5c227 : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbd1044676804ece878eeab85dad951f : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54d842674d654db4a05f574e72ca0d56 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67028d0cdb504e4e85918a1037bed620 : SUCCESS in 2m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d2b2e9fd3df48e9a15a00224cbb0ecb : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ae61a4ae89433ab7221e25b5c6d5c2 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c22dd2e7db954049a73733150ecc4ae3 : SUCCESS in 4m 46s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/b62bdb41885d4edeb22cadf07549d247 : SUCCESS in 41m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d1ed3a5db4d4cec49928007885720cdd43121cc3","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-03 09:16:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"03f2e31f0bd9f2991de96432b0409091e05e3eb9","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2020-01-03 09:23:00.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"76ffdf14c6b9fa4ff723a0eecb6fdb544cd3d724","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2020-01-06 06:43:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f632732b79bf560d406b2e32f43ace025876e03d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 06:44:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ae62bb38dfe504e69c2ddc528d1f784ae3cc8a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 06:56:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"48486ac77b759e3f9948d658867cd2f801e5b5cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 06:56:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/59d78c7160954a1686c256b2b6fca349 : SUCCESS in 3m 02s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/cfb392d6783e49fda387e0e5c909c6f1 : SUCCESS in 5m 28s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/7b11f9f0990f43a5afafbd2a11e35ede : SUCCESS in 6m 43s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/b9879abcbb684f73a69d7ca2e243f438 : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecd5bf3f12d0466ea499c33e5981a782 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e464ad7b4134d6186258647e93bd34a : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/260e5129a5bb4e558db8ffb209536b11 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07ce990d0cb3448782814562c96bbd0a : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0734b10a159940239f924a91dce00aa3 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd2ceaa804934db59d9d29f4085d84ca : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":2},{"id":"c4746806f8a62cb2794b36ff2b72df8cd220b38d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 06:59:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebeb91afe200446086ec0ffbb57d8930 : SUCCESS in 1m 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ef79b536ea1490f92388b394290b150 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"47561ba1e3997daca1b64d99f2a1b7d88619488f","revisions":{"1314ef00c48d0b47b790af7d56e3ea2d4b35569a":{"kind":"REWORK","_number":1,"created":"2020-01-01 20:35:59.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/45/700845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/45/700845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/45/700845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/45/700845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/45/700845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/45/700845/1"}}},"commit":{"parents":[{"commit":"70a293d31ac4fb7f465b59e7a3c539c5629a205f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/70a293d31ac4fb7f465b59e7a3c539c5629a205f"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-01 20:35:11.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-01 20:35:57.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2. requirements.txt and\nlower-constraints.txt are updated accordingly. for more info. please\nrefer [1].\nDepends-On: https://review.opendev.org/#/c/700733/\n[1] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: I29cd34d17a00eb1b576f46d6010203603b5c8165\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/1314ef00c48d0b47b790af7d56e3ea2d4b35569a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/1314ef00c48d0b47b790af7d56e3ea2d4b35569a"}]},"branch":"refs/heads/master"},"47561ba1e3997daca1b64d99f2a1b7d88619488f":{"kind":"REWORK","_number":2,"created":"2020-01-02 19:05:27.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/45/700845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/45/700845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/45/700845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/45/700845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/45/700845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/45/700845/2"}}},"commit":{"parents":[{"commit":"70a293d31ac4fb7f465b59e7a3c539c5629a205f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/70a293d31ac4fb7f465b59e7a3c539c5629a205f"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-01 20:35:11.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-02 19:05:13.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2. requirements.txt and\nlower-constraints.txt are updated accordingly. for more info. please\nrefer [1].\nDepends-On: https://review.opendev.org/#/c/700733/\n[1] https://etherpad.openstack.org/p/Enable_Django22_support_in_Horizon_Plugin\n\nChange-Id: I29cd34d17a00eb1b576f46d6010203603b5c8165\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/47561ba1e3997daca1b64d99f2a1b7d88619488f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/47561ba1e3997daca1b64d99f2a1b7d88619488f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
