)]}'
{"id":"openstack%2Fironic-ui~701686","triplet_id":"openstack%2Fironic-ui~stable%2Fstein~I9a310446743cfe6bca5480455b91a7cd74da8785","project":"openstack/ironic-ui","branch":"stable/stein","topic":"requirements-fix","hashtags":[],"change_id":"I9a310446743cfe6bca5480455b91a7cd74da8785","subject":"Fix horizon dependency","status":"MERGED","created":"2020-01-09 08:38:17.000000000","updated":"2020-01-17 09:39:48.000000000","submitted":"2020-01-17 09:36:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701686-1579253765860-a981c286","meta_rev_id":"54751d2bedba3d83002547d7e2201db605302add","_number":701686,"virtual_id_number":701686,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-17 09:36:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-01-17 09:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-01-17 09:23:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 14:35:11.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-01-17 09:23:42.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-01-17 09:36:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff22d3df083e999bfe1e632c2ecf7c7b45c3a36d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-09 08:38:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f607cd1187dcf9ac2c43b3dd091c003095556e6","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-09 08:39:02.000000000","message":"Topic set to requirements-fix","accounts_in_message":[],"_revision_number":1},{"id":"5b9e5a32e7e9bf1ce2649077a2b8dbc3c6fbda6b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-09 09:42:21.000000000","message":"Patch Set 1:\n\nmmmm pretty sure this is not compatible with global-requirements","accounts_in_message":[],"_revision_number":1},{"id":"93a53ebd5ad3cd4e39a66ef37588eb14d0aac3b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 09:45:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f91459ea6b41463cbf747e378acb53b9 : FAILURE in 3m 20s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/5ecd37cc27aa463c8ad2e7e2041324cd : SUCCESS in 6m 39s\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/ba1e928a2f124d21bf9cd6753bafa937 : SUCCESS in 11m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/539373223e784e17b6b8976576decf01 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd170a39df6d483fa6c4b6f3e74a8f10 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7537f24159594985afd926c0fb01983a : SUCCESS in 6m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4a5981108bf94daba0f0b19b91a1f098 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/527feb0d76104d38ac63f18604102802 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e75803c230744babab035b9a6d68077 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d06bb84873e6494b9535ac8f369aaed2 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19a80f8c0ce24314bc8cee349958cce6 : SUCCESS in 6m 06s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/75a65758c91b41d38e4909e23e463210 : SUCCESS in 44m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50db9e574b07cedb21b141f5bd410b3484b1821a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-09 11:07:02.000000000","message":"Patch Set 1: Workflow-1\n\nWIP till stable/train change lands.","accounts_in_message":[],"_revision_number":1},{"id":"b54f987445fcf8c4f1eeeabcbf811b9adb1ae40e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-09 11:52:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f48d0beb0980ef5eefd4256ae361efa3c55d4e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 12:30:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/253acafa0af44e05bda7fe8f65796880 : SUCCESS in 2m 38s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/66fd3e4c5cf44dfbbb5b39601ea9b075 : SUCCESS in 6m 05s\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/dde160b726a64cc1845cb73984297142 : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61d2e0ae1df34d13a8ad535b5abbdb28 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f7329c26375411c9da18730cab0fdd5 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0cc8140a05f64bd89caebd697802bfbe : SUCCESS in 3m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f2d64a7926d842429527b88eb75184e1 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e683e524afc948438ab5b204128d494c : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afe56c31e5374ab784dbd0897a997dcf : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f4d9071dd564a82aadf9844b67def5a : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8da840011c0e478b9510175a3aae0948 : SUCCESS in 4m 51s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/fcd573aafb0344799a36b09781624738 : SUCCESS in 36m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad00f3c728995df7f14631a9e3f96ac530cbf294","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-10 09:46:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6dd1250e201f795f706eda0b132d2fda7633e860","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-10 14:35:11.000000000","message":"Patch Set 2: Code-Review+1\n\nlooks okay to me, need upper branch to merge first.","accounts_in_message":[],"_revision_number":2},{"id":"e38d69dbba5e678e5baa3bfa77ff76da5855067b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-11 16:10:19.000000000","message":"Patch Set 2: Workflow-1\n\nI will update this once the stable/train patch lands.","accounts_in_message":[],"_revision_number":2},{"id":"8408c24fe3958ee92cdaf8441eed12a03f555ff3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-14 23:14:33.000000000","message":"Patch Set 2: Code-Review-1\n\nlower-constraints job does not check lower bound.... For example, Django lower bound is 1.11 but Django 2.0.13 is installed [1]. This happens because install_command is not configured as expected and pip install command takes too -c arguments [2]. Once stable/train patch lands I will cherry-pick it again.\n\n[1] https://zuul.opendev.org/t/openstack/build/61d2e0ae1df34d13a8ad535b5abbdb28/log/tox/lower-constraints-1.log#249\n[2] https://zuul.opendev.org/t/openstack/build/61d2e0ae1df34d13a8ad535b5abbdb28/log/tox/lower-constraints-1.log#3","accounts_in_message":[],"_revision_number":2},{"id":"f7dd1afc47ffb9512a4c6a219ffc4b67c44efbed","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-16 07:49:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"91dcd05515f59e622a1e2e9df5c717de7c6bf51e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:29:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/383087d7d82f4b2f90b84c5c85fee227 : SUCCESS in 3m 07s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/2035e56bcb7f4170970b7dfc49e52970 : RETRY_LIMIT in 4m 29s\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/6450b40c64e5472dbdbf130b62e816a2 : RETRY_LIMIT in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce949f2cf3674882886787d4fb2a46fc : FAILURE in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceabb38e1d434be7bf973d4078f0147b : FAILURE in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47674cc28afc4e30845e934af185305c : FAILURE in 2m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/50be2fd1bc3444d8a3a4128bbd189120 : FAILURE in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11f590cfddaa4690bd44fa4df0672fda : FAILURE in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04d23c2876be45aaac4ce889557ddcd3 : FAILURE in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb3db16dc034a7385b844b5404f6806 : FAILURE in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a150651331a4db39038a07748fdb302 : SUCCESS in 5m 55s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/47314964d47f44dd84dfb310085d350d : FAILURE in 33m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cb87ca4594045ca69fda28d05dd02a63d928786c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-16 09:13:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"247470a492164c4eb9d12f08f55fdfacfe8b5c64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 09:56:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/eb011412459443aaabf80ef52562a791 : SUCCESS in 3m 47s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/39fb627fdea5465aaf77dcda69e0bad5 : SUCCESS in 6m 20s\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/8efd17f4ba2b4b07a9f12cfaceda0a2d : SUCCESS in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f865777f029e46d8bdbdf22dceb7832c : FAILURE in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f083a4a5a2334063aaf149cba651c89d : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6b55dbb6b8a40faac51a7a1da35a17d : SUCCESS in 3m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a711f408ec4c44839f5075f6e86f6e6d : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e84714d7937b4c1fbf1f1f8773ba6d2f : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ace9390f62e34fe9889393b4e9b3c52e : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29d2f73c42534185acf2cc8709407322 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8a6099ff6bd416fb55957eed1da753b : SUCCESS in 6m 04s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/cb1f34cd9bb24240b859c416b69ee0fd : SUCCESS in 41m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c3d818b8e72f89c6194f9033e442dc3d7255b23f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-16 16:19:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b0e76e8e1910935d45e68fe33068940a10229821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 17:07:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a46ae96e556d44b8b81abf0d6e578f06 : SUCCESS in 2m 41s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/ce2f0332a3f944b2b83af22e9c66f59d : SUCCESS in 7m 24s\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/075ff368bfa142e0b3f2f5852140ec54 : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7245a6006bf24045b9d363b3098bcc74 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/934dfdc041d84a9ab5edb8bf74106111 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bfda023300fb4fa7b023444af5faf1f4 : SUCCESS in 4m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5b2e955c27864fa9ad9fd18944088c08 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8aa03c4915f944d98919dd63869358e2 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/677b19d93b7540019473c2c08054ebad : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd41305b34ff4eb1853885391d3d1162 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0db8f6aca4e4c2eb9cff9bdcf76495d : SUCCESS in 6m 15s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/4b1aa5a4a1ff4cecbe3db3dd4005ce0d : SUCCESS in 37m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"63db19780685abe7f9d69e5da63092ec304b7bb6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-17 09:23:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nthanks!","accounts_in_message":[],"_revision_number":5},{"id":"36ef059523fdf6f197c95d4120b24e02ca052217","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 09:23:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f6d756d1cfa059b60a426c4e3fd67f2752a69d9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 09:36:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ce0e0e16d6e44ddd93f324fafe76ad7e : SUCCESS in 5m 14s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/49a1f72c541746d79ad7bed0e2e2e442 : SUCCESS in 7m 42s\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/6a5ab87ef4da4210b5a04b0b5c0a81d7 : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e87629a83e5b4b2887dd0644a5baf884 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea5c521142b47a5a0e3beca75ba5e28 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18fb7b056e3b415e973db2212240be4e : SUCCESS in 6m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/873c75cc920b468cb2a1f97fe55c9bd0 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/029d898aa90644f4bc7b9adda9af85a0 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a58ed3f0373a4bec958e38e7e70335d3 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c54b18751f24e47a1ec1a7884d8a33e : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5952a291025455c82ac6f3545ba69c4 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":5},{"id":"26ef12a6bd4889fd12e18e83b7d305c84ab0b9db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 09:36:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"54751d2bedba3d83002547d7e2201db605302add","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 09:39:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7439b3c61f041158e91a9b485621c4d : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e91e8ed5ad7d48c0916f27f7fe91b36e : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5c593fd34a149c62e4153a1456bb5bc1ae366ea7","revisions":{"9ccf95cb1a0575df7a737d7730b7d66da7682e6c":{"kind":"REWORK","_number":1,"created":"2020-01-09 08:38:17.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/86/701686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/86/701686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/1"}}},"commit":{"parents":[{"commit":"21167c690e879f36c1ce321729786c6409c18945","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/21167c690e879f36c1ce321729786c6409c18945"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-09 08:30:20.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-09 08:37:56.000000000","tz":540},"subject":"Fix horizon dependency","message":"Fix horizon dependency\n\nThis commit moves horizon to requirements.txt.\n\nhorizon was added to uper-constraints in stable branches recently,\nso installing from the tarball no longer works. upper-constraints\nin stable branches picks up the right version of horizon, so there\nis no need for the upper bound of the horizon version.\n\nIn addition, the link of the horizon tarball was not updated after\nthe stable branch was created. I don\u0027t know why it works for stable/train\nbut I believe this commit mitigates the situation as the upper-constraints\nfile in the requirements repo picks up the right version of horizon.\n\nThis commit is proposed directly as the similar change in the master\nbranch happened as part of Django 1.11 drop.\n\nChange-Id: I9a310446743cfe6bca5480455b91a7cd74da8785\n(cherry picked from commit 50475021dbe9a643043cdb39f57c01aede008dbb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/9ccf95cb1a0575df7a737d7730b7d66da7682e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/9ccf95cb1a0575df7a737d7730b7d66da7682e6c"}]},"branch":"refs/heads/stable/stein"},"9069658480a6ccd58f0d462033c86ece366820a6":{"kind":"REWORK","_number":2,"created":"2020-01-09 11:52:28.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/86/701686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/86/701686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/2"}}},"commit":{"parents":[{"commit":"21167c690e879f36c1ce321729786c6409c18945","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/21167c690e879f36c1ce321729786c6409c18945"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-09 08:30:20.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-09 11:51:57.000000000","tz":540},"subject":"Fix horizon dependency","message":"Fix horizon dependency\n\nThis commit moves horizon to requirements.txt.\n\nhorizon was added to uper-constraints in stable branches recently,\nso installing from the tarball no longer works. upper-constraints\nin stable branches picks up the right version of horizon, so there\nis no need for the upper bound of the horizon version.\n\nIn addition, the link of the horizon tarball was not updated after\nthe stable branch was created. I don\u0027t know why it works for stable/train\nbut I believe this commit mitigates the situation as the upper-constraints\nfile in the requirements repo picks up the right version of horizon.\n\nThis commit is proposed directly as the similar change in the master\nbranch happened as part of Django 1.11 drop.\n\nstable/stein only:\nsphinx entry in test-requirements.txt did not match global-requirements.\nThis commit updates it to match the current global-requirements.\n\nChange-Id: I9a310446743cfe6bca5480455b91a7cd74da8785\n(cherry picked from commit 864c78b30cb740ac4e2cd0eb42e846215c64fc18)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/9069658480a6ccd58f0d462033c86ece366820a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/9069658480a6ccd58f0d462033c86ece366820a6"}]},"branch":"refs/heads/stable/stein"},"10cee0a3f55b1983341c0aa4284ad425f8ebdece":{"kind":"REWORK","_number":3,"created":"2020-01-16 07:49:49.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/86/701686/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/86/701686/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/3"}}},"commit":{"parents":[{"commit":"21167c690e879f36c1ce321729786c6409c18945","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/21167c690e879f36c1ce321729786c6409c18945"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-09 08:30:20.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-16 07:35:46.000000000","tz":540},"subject":"Fix horizon dependency","message":"Fix horizon dependency\n\nThis commit moves horizon to requirements.txt.\n\nhorizon was added to upper-constraints in stable branches recently,\nso installing from the tarball no longer works. upper-constraints\nin stable branches picks up the right version of horizon, so there\nis no need for the upper bound of the horizon version.\n\nIn addition, the link of the horizon tarball was not updated after\nthe stable branch was created. I don\u0027t know why it works for stable/stein\nbut I believe this commit mitigates the situation as the upper-constraints\nfile in the requirements repo picks up the right version of horizon.\n\nThis commit is proposed directly as the similar change in the master\nbranch happened as part of Django 1.11 drop [1].\n\nIn addition, ironic-ui tox.ini specified install_command with the\nupper-constraints.txt. The contraints file is now specified in \"deps\"\nsection to follow the convention used widely in the OpenStack community.\ndoc/requirements.txt is cleanup and the runtime and test requirements\nare now added to \"deps\" of the \"docs\" env to generate the module index\nproperly.\n\nstable/stein only:\nsphinx entry in test-requirements.txt did not match global-requirements.\nThis commit updates it to match the current global-requirements.\n\n[1] https://review.opendev.org/#/c/700845/\n\nChange-Id: I9a310446743cfe6bca5480455b91a7cd74da8785\n(cherry picked from commit efa4927bb7baeb918fad3734b21b0223e0958404)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/10cee0a3f55b1983341c0aa4284ad425f8ebdece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/10cee0a3f55b1983341c0aa4284ad425f8ebdece"}]},"branch":"refs/heads/stable/stein"},"6a1395cec208342efcdc5a565a20a8b26f49c209":{"kind":"REWORK","_number":4,"created":"2020-01-16 09:13:11.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/86/701686/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/86/701686/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/4"}}},"commit":{"parents":[{"commit":"21167c690e879f36c1ce321729786c6409c18945","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/21167c690e879f36c1ce321729786c6409c18945"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-09 08:30:20.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-16 09:12:59.000000000","tz":540},"subject":"Fix horizon dependency","message":"Fix horizon dependency\n\nThis commit moves horizon to requirements.txt.\n\nhorizon was added to upper-constraints in stable branches recently,\nso installing from the tarball no longer works. upper-constraints\nin stable branches picks up the right version of horizon, so there\nis no need for the upper bound of the horizon version.\n\nIn addition, the link of the horizon tarball was not updated after\nthe stable branch was created. I don\u0027t know why it works for stable/stein\nbut I believe this commit mitigates the situation as the upper-constraints\nfile in the requirements repo picks up the right version of horizon.\n\nThis commit is proposed directly as the similar change in the master\nbranch happened as part of Django 1.11 drop [1].\n\nIn addition, ironic-ui tox.ini specified install_command with the\nupper-constraints.txt. The contraints file is now specified in \"deps\"\nsection to follow the convention used widely in the OpenStack community.\ndoc/requirements.txt is cleanup and the runtime and test requirements\nare now added to \"deps\" of the \"docs\" env to generate the module index\nproperly.\n\nstable/stein only:\nsphinx entry in test-requirements.txt did not match global-requirements.\nThis commit updates it to match the current global-requirements.\n\n[1] https://review.opendev.org/#/c/700845/\n\nChange-Id: I9a310446743cfe6bca5480455b91a7cd74da8785\n(cherry picked from commit efa4927bb7baeb918fad3734b21b0223e0958404)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/6a1395cec208342efcdc5a565a20a8b26f49c209"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/6a1395cec208342efcdc5a565a20a8b26f49c209"}]},"branch":"refs/heads/stable/stein"},"5c593fd34a149c62e4153a1456bb5bc1ae366ea7":{"kind":"REWORK","_number":5,"created":"2020-01-16 16:19:44.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/86/701686/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/86/701686/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/86/701686/5"}}},"commit":{"parents":[{"commit":"21167c690e879f36c1ce321729786c6409c18945","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/21167c690e879f36c1ce321729786c6409c18945"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-09 08:30:20.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-16 16:17:44.000000000","tz":540},"subject":"Fix horizon dependency","message":"Fix horizon dependency\n\nThis commit moves horizon to requirements.txt.\n\nhorizon was added to upper-constraints in stable branches recently,\nso installing from the tarball no longer works. upper-constraints\nin stable branches picks up the right version of horizon, so there\nis no need for the upper bound of the horizon version.\n\nIn addition, the link of the horizon tarball was not updated after\nthe stable branch was created. I don\u0027t know why it works for stable/stein\nbut I believe this commit mitigates the situation as the upper-constraints\nfile in the requirements repo picks up the right version of horizon.\n\nThis commit is proposed directly as the similar change in the master\nbranch happened as part of Django 1.11 drop [1].\n\nIn addition, ironic-ui tox.ini specified install_command with the\nupper-constraints.txt. The contraints file is now specified in \"deps\"\nsection to follow the convention used widely in the OpenStack community.\ndoc/requirements.txt is cleanup and the runtime and test requirements\nare now added to \"deps\" of the \"docs\" env to generate the module index\nproperly.\n\nstable/stein only:\nsphinx entry in test-requirements.txt did not match global-requirements.\nThis commit updates it to match the current global-requirements.\n\n[1] https://review.opendev.org/#/c/700845/\n\nChange-Id: I9a310446743cfe6bca5480455b91a7cd74da8785\n(cherry picked from commit efa4927bb7baeb918fad3734b21b0223e0958404)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/5c593fd34a149c62e4153a1456bb5bc1ae366ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/5c593fd34a149c62e4153a1456bb5bc1ae366ea7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
