)]}'
{"id":"openstack%2Fironic-ui~701698","triplet_id":"openstack%2Fironic-ui~master~I329e43d51aaeade8d4535b2f68cec7680b5ebfce","project":"openstack/ironic-ui","branch":"master","topic":"tox-enforce-correct-pyver","hashtags":[],"change_id":"I329e43d51aaeade8d4535b2f68cec7680b5ebfce","subject":"Enforce running tox with correct python version based on env","status":"MERGED","created":"2020-01-09 09:31:53.000000000","updated":"2020-01-09 11:27:57.000000000","submitted":"2020-01-09 11:25:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701698-1578569153830-893a25b6","meta_rev_id":"089f121f2ef6775c7ecec891a062f7c9073793e3","_number":701698,"virtual_id_number":701698,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 11:25:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2020-01-09 10:58:13.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2020-01-09 11:00:09.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":1,"date":"2020-01-09 11:00:09.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 10:58:13.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-01-09 11:00:09.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-01-09 11:25:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f2de3a1decf120880c72ab79edd95e89bc6c540","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-09 09:31:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf8c38e9fd14621c3f5cb26d4480175f2e375671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 10:21:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/bc48aaa135c34740986b39f078c53591 : SUCCESS in 5m 06s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/e2838cb74120462aa506a8d2e5600ddd : SUCCESS in 6m 52s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/36fdcd380f9c4819bd4edcaf6a41b85c : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ace974a939e4815a99bd7915ecbb501 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/998e1e73006b41d7aa29a4ab7ea52bff : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2650c53e7a99450c865d9268a5f5eb12 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d68723fade9942e9a3121874b54fbf8f : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c0172ea78741fa8693523d2fd628e1 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb14b6f209984ea8beb4c38289990df1 : SUCCESS in 5m 39s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/48655e95ec914f23962ce6fd0103fd99 : SUCCESS in 43m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df459d0e9cf442a19ba0916148ee5c627ed9c9ec","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-01-09 10:58:13.000000000","message":"Patch Set 1: Code-Review+2\n\nThank you Riccardo.","accounts_in_message":[],"_revision_number":1},{"id":"657684730d81dd3234a7c758b4043d1c29728b7f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-09 11:00:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis makes sense and we do the same thing in most repos.","accounts_in_message":[],"_revision_number":1},{"id":"ff82b30baab438fb7aeb3e4ecb4e83d1bfbdd83c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 11:00:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e0a3f1520355770a8117dd3b633614710d6d679","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 11:25:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bba56f56c79e341048182536f566b569c0ef314b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 11:25:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/225e5a47c70a4912a982facff8f4e7be : SUCCESS in 6m 50s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/02fd9238040142c484864ee47b05dd47 : SUCCESS in 6m 08s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/4f88bf92fa004d2f98bccf86967c7804 : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a198be2f815044e7a098ee658605bec9 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd367f84e05540d888faf65a5a4141fd : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88a8aa415488419d8e8fa535215a1c7c : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02e7abe9ae1b4f0a882dee6cb0106734 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e849b81ed2645e0ac38e9b263ac6af9 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/658b58af485542898055c4c1969c580b : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"16bf87c02473ddfbe500cfdffd8bc097c9965113","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 11:27:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27eadc7eed646d39689e8ec182dafaa : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c67b591c1144081a1ed83be6bf005c8 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31234a301dd69a581ed414ec12606e769bbadc72","revisions":{"31234a301dd69a581ed414ec12606e769bbadc72":{"kind":"REWORK","_number":1,"created":"2020-01-09 09:31:53.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/98/701698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/98/701698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/98/701698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/98/701698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/98/701698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/98/701698/1"}}},"commit":{"parents":[{"commit":"f92ab9c7ff6d5910aad59b8df1d7f8a0ce7251c3","subject":"Merge \"Drop Django 1.11 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/f92ab9c7ff6d5910aad59b8df1d7f8a0ce7251c3"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-09 09:31:20.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-09 09:31:51.000000000","tz":60},"subject":"Enforce running tox with correct python version based on env","message":"Enforce running tox with correct python version based on env\n\nSince removing support for Python 2, we changed the basepython\nvalue to 3.\nThis means that all the tox tests run with the default python\nversion available in the system.\nThis is not quite correct when running on environment such as\npy36, py37 or py38, since they imply running with different\nPython versions based on the environment.\nTo enforce the correct version we need to add the option\nignore_basepython_conflict available since tox 3.1.0 [0].\n\n[0] https://tox.readthedocs.io/en/latest/config.html#conf-ignore_basepython_conflict\n\nChange-Id: I329e43d51aaeade8d4535b2f68cec7680b5ebfce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/31234a301dd69a581ed414ec12606e769bbadc72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/31234a301dd69a581ed414ec12606e769bbadc72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
