)]}'
{"id":"openstack%2Fironic-ui~735028","triplet_id":"openstack%2Fironic-ui~master~I76c9f19bb11c9d5a92a06e3b7c8b70b7c3a035e5","project":"openstack/ironic-ui","branch":"master","hashtags":[],"change_id":"I76c9f19bb11c9d5a92a06e3b7c8b70b7c3a035e5","subject":"api: drop project_id from kwargs in client","status":"MERGED","created":"2020-06-11 01:22:28.000000000","updated":"2020-11-02 11:48:17.000000000","submitted":"2020-06-11 20:22:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735028-1591906978018-2acfc7a0","meta_rev_id":"eac0c37acd33b3e0d92d411ef37b7e3cd3590c44","_number":735028,"virtual_id_number":735028,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 20:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-06-11 20:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2020-06-11 20:08:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-11 20:08:04.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-06-11 20:22:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0610c45a489f8a55949edf0790f4f83fb456a531","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-11 01:22:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de144aad664125e2f19fe1f0dd23ff1ddab95a2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 02:03:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/80d9630c567143a09947b21e3224b8eb : SUCCESS in 4m 34s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/801ad51282694a34a312c34993c8eef6 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed023950de0a4d2fb003e9755a99a30b : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06f49cb9b1c34bd8a27f75d75ef35a0e : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d48c18c790492f99b9e810ab8cf500 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4ea0131b9c0491b9a1507f435eb57ae : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/485399bbd4194841af0bb43f9d77fb62 : SUCCESS in 8m 02s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/01b510e4a25b49689dd734907c69cca0 : SUCCESS in 39m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb4c9410382015955c54f87fa6174d2732c140f9","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-06-11 20:08:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMakes sense since the user would already have a token. This will likely need to be changed at some point, but I\u0027d prefer a user to report that it is not working in their case first.\n\n 1:02:54 PM \u003cTheJulia\u003e mnaser: is it actually in the client library that the failure is happening, because it seems like project id is valid to pass through for non-admins if the rights are correct\n1:03:02 PM \u003cTheJulia\u003e then again I\u0027m not sure people are actually setting that\n1:03:40 PM \u003cmnaser\u003e TheJulia: ok so based on my troubleshooting -- https://github.com/openstack/python-ironicclient/blob/master/ironicclient/client.py#L51-L60\n1:04:00 PM \u003cmnaser\u003e so in our case, auth_type is not provided, endpoint and token is set, so auth_type is set to admin_token\n1:04:18 PM \u003cmnaser\u003e which takes us to the following plugin - https://github.com/openstack/keystoneauth/blob/master/keystoneauth1/loading/_plugins/admin_token.py\n1:05:03 PM \u003cmnaser\u003e which only needs endpoint and a token to \u0027be loaded\u0027, hence i would get an error that project_id was provided but not needed (and really, if we have a token and an endpoint, we really dont need a project anyways)\n1:05:46 PM \u003cmnaser\u003e maybe the real bug is that the client shouldn\u0027t be using admin_token auth_type but that\u0027s up for debate/discussion\n1:06:14 PM \u003cmnaser\u003e git blame in the python ironicclient says that code has been there for 4 years so\n1:06:15 PM — mnaser shrugs\n1:06:17 PM \u003cTheJulia\u003e figures it would ahve been one of my changes","accounts_in_message":[],"_revision_number":1},{"id":"0342235586a5bdc9cd158c06e22596513fa00b44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 20:08:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e06cd5ae0bbd0516f8d23c78879de66dd62302be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 20:22:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/014f30ef0a814003880ae06aa280f1dd : SUCCESS in 7m 32s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/97eb5306ecac490d9871fd812a4ab852 : SUCCESS in 11m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e3332e3c4c949c6bb0190b14d8fad3d : SUCCESS in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14a4c9a3dbb145698f65356b72a5eed8 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/930b41ef42624871bd78231632e8a44e : SUCCESS in 9m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/353729ba728d4fa9af84862e5f61372c : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35faea0021f4a5a82cfc70d747180ff : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":1},{"id":"77145851e907d1801fcad15b7b7da81ddf37b161","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 20:22:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"65c659c233442735ed64b95ea5aec8f3ff53121e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 20:25:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14ba16c8dbd4473dbba22ee22d33c0c1 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":1},{"id":"eac0c37acd33b3e0d92d411ef37b7e3cd3590c44","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-11-02 11:48:17.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 0e2f805468210169c98b87d7bbf7db60b8a3e29e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50134dc769a666b29ba55e377d7b1bfc4af94489","revisions":{"50134dc769a666b29ba55e377d7b1bfc4af94489":{"kind":"REWORK","_number":1,"created":"2020-06-11 01:22:28.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/28/735028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/28/735028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/28/735028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/28/735028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/28/735028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/28/735028/1"}}},"commit":{"parents":[{"commit":"6d253ca7f1542c758dc1d55074fc5fba13060153","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/6d253ca7f1542c758dc1d55074fc5fba13060153"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-06-11 01:21:02.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-06-11 01:22:26.000000000","tz":-240},"subject":"api: drop project_id from kwargs in client","message":"api: drop project_id from kwargs in client\n\nThe client will use the admin_token auth_type if an endpoint and token\nare provided, this auth_type does not support getting a project_id so it\nwill raise an error in the dashboard.\n\nChange-Id: I76c9f19bb11c9d5a92a06e3b7c8b70b7c3a035e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/50134dc769a666b29ba55e377d7b1bfc4af94489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/50134dc769a666b29ba55e377d7b1bfc4af94489"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
