)]}'
{"id":"openstack%2Fironic-ui~760895","triplet_id":"openstack%2Fironic-ui~stable%2Fussuri~I76c9f19bb11c9d5a92a06e3b7c8b70b7c3a035e5","project":"openstack/ironic-ui","branch":"stable/ussuri","hashtags":[],"change_id":"I76c9f19bb11c9d5a92a06e3b7c8b70b7c3a035e5","subject":"api: drop project_id from kwargs in client","status":"ABANDONED","created":"2020-11-02 11:48:17.000000000","updated":"2021-01-13 10:00:50.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e10b9f6d60b0b2ca2929e417aadb3cbfefa0fe16","_number":760895,"virtual_id_number":760895,"owner":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-02 12:50:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"date":"2020-12-22 09:39:44.000000000","_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":-1,"date":"2021-01-11 14:54:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-11-02 11:48:17.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-02 11:48:17.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-11-02 12:50:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 05:21:07.000000000","updated_by":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"fcd1ea07e80e3285284b5c52a822ac01d6f4ef82","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-11-02 11:48:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"84f0a76f16538528f974571afef4475ae107fe32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 12:50:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/f1d6556b37194da1afbcf1f8aeba8720 : SUCCESS in 5m 07s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/0ecab9abb1f74961a91979223a6df1b7 : FAILURE in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d707979a73b4ed3864da24e6a5072e7 : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d045002498de4ad196ab9e2526edfcd1 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87880f44d13d41d9bfb1bc383934c23c : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/581d4630dadb40ec9adc8d72280a5a44 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a58c13b2671c45feaaa2c5bab6b7ad78 : SUCCESS in 6m 05s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/9b9b055fc7bb456c96cddaba167f25a1 : SUCCESS in 43m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9dbe01623ce129f740569382fc586cf2677fff3","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-11-10 13:39:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b8fb927ba9f49fb170984f3ff53818da7545997","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 14:39:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/f3f6978c78314108a0ca9f6b4075356e : SUCCESS in 5m 54s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/d88f6a04f94544b08af1bafaed764740 : FAILURE in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52c29a3bc58749fe861b8c1b0643f1bf : SUCCESS in 16m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5494e934e38f4e45baf1cc137dcee870 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb81175765e4a379aab154d2e576f04 : SUCCESS in 11m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f809d8f87d5643d8ba2bf52b69b436fb : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ddb05a86be5453fbca440d7b00306c0 : SUCCESS in 23m 53s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/4272278f84f745c8aefeeb2885ae005d : SUCCESS in 53m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cbf95b27a2376fdc1d8286305d8f238eebab28eb","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-11-13 12:14:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45ecfb67e3ff63e32dc91427711c9e4d7631689c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 12:58:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/5b913c715c514b5083ac0f18b2ade822 : SUCCESS in 4m 30s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/8cb6194ceb184147a1e1356e989fd1f7 : FAILURE in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d57546f56dee476285c7d15df73aa228 : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/152eb75c3d2145cea6da42a618a97b1c : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c0bd1d2952e4cd6a20b5a5d47c305b3 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0358e322e1846ddbabc6397c558fac6 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d953e4b1de34e428753093a4e5c652b : SUCCESS in 5m 46s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/af9c8c9134f14f62a153f615fcfb16ff : SUCCESS in 43m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0577a714311f2f588b990538500cba0a30d7424f","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-11-24 05:21:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f91728c1a4a289e39be44338043f870763ab4122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 06:07:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/3b9f04fa05604d7ebaa5c974eb104cdb : SUCCESS in 4m 43s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/9a6a91c05c964219af1bfef73741bf94 : FAILURE in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/948d8b0974e748ef8a540c7fa0579d24 : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0157d535ce6f40f0b5ec5698e285ceef : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a927fa1396e4b859de587cf9477e9b0 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94cc64ceaaa2444ca4662c8c72952d80 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98201629fdd4919962d058b376d86aa : SUCCESS in 6m 18s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/ffc1fb52b6644e92997cd0983d6c6db9 : SUCCESS in 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0f1fe5bd276e45f76184d810f2ef92cc7eabeee","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-22 09:39:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc386eb14d0345f84476cdf35690761f911a57ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 10:22:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/2d3c2156f9fb48b186e885ce16231353 : SUCCESS in 5m 22s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/abb75348aa6243e2b3c643fa1cab9d6c : FAILURE in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d99edcb6a03c48489ae1c85b287f37a1 : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2df1a199bc864f83ad79ab8bf3061cf0 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9c5c9f525b94bb594647f3d88db2ac8 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d97b5180aa74ed18e0122e2dc056ab7 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c8025a87254ae78579fb9e933fb280 : SUCCESS in 6m 48s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/63342849ec1047918c1abd975415e554 : SUCCESS in 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8292bf3a7508352cd521553870825996aab0645d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-11 14:54:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e10b9f6d60b0b2ca2929e417aadb3cbfefa0fe16","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2021-01-13 10:00:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e2f805468210169c98b87d7bbf7db60b8a3e29e","revisions":{"0e2f805468210169c98b87d7bbf7db60b8a3e29e":{"kind":"REWORK","_number":1,"created":"2020-11-02 11:48:17.000000000","uploader":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"ref":"refs/changes/95/760895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/95/760895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/95/760895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/95/760895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/95/760895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/95/760895/1"}}},"commit":{"parents":[{"commit":"feb6e927384259b3d14d34cfa60d5ddf9a89d820","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/feb6e927384259b3d14d34cfa60d5ddf9a89d820"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-06-11 01:21:02.000000000","tz":-240},"committer":{"name":"dou","email":"douyali@troila.com","date":"2020-11-02 11:48:17.000000000","tz":0},"subject":"api: drop project_id from kwargs in client","message":"api: drop project_id from kwargs in client\n\nThe client will use the admin_token auth_type if an endpoint and token\nare provided, this auth_type does not support getting a project_id so it\nwill raise an error in the dashboard.\n\nChange-Id: I76c9f19bb11c9d5a92a06e3b7c8b70b7c3a035e5\n(cherry picked from commit 50134dc769a666b29ba55e377d7b1bfc4af94489)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/0e2f805468210169c98b87d7bbf7db60b8a3e29e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/0e2f805468210169c98b87d7bbf7db60b8a3e29e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
