)]}'
{"id":"openstack%2Fironic-ui~766880","triplet_id":"openstack%2Fironic-ui~master~Ic7094ab6ed01af2bcbe0beb13f300499215fd903","project":"openstack/ironic-ui","branch":"master","topic":"remove-l-c-job","hashtags":[],"change_id":"Ic7094ab6ed01af2bcbe0beb13f300499215fd903","subject":"Remove Chrome browser from karma.conf.js","status":"MERGED","created":"2020-12-14 05:26:39.000000000","updated":"2020-12-18 14:00:07.000000000","submitted":"2020-12-18 13:55:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766880-remove-l-c-job","meta_rev_id":"4491fd420cacb468c30d78e4b978aa103ce5817d","_number":766880,"virtual_id_number":766880,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 13:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":1,"date":"2020-12-16 21:51:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-12-16 10:18:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2020-12-18 13:32:49.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":1,"date":"2020-12-18 13:32:49.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-14 06:00:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-14 14:44:46.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-12-16 10:18:06.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-12-16 21:51:51.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2020-12-18 13:32:49.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"}],"messages":[{"id":"af5d187863b3d6cd83eafac70611747bd63d2874","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-14 05:26:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bef48e5d34d010c2fe7cc7a9252b7f830e3dd71b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 06:00:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/96a9e2ef35a34e1da42d87f485ade61f : SUCCESS in 2m 28s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/5b1726ae71ea44ec8d22d325558112e4 : SUCCESS in 4m 31s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/c2ddf867c1c54017ae657e88ef8f71b3 : FAILURE in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3e13aed88494d7ea8c9d2b123f50906 : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8fcd8a297f0478f9b1bd34a03d1a2c2 : FAILURE in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/871cb56476e840ba9ce31c2928585d07 : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25193dab817440eaa7d2d429f8aa0d2e : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea0451d4720f4896b4eae460a6170718 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957673e670cf4b398d85e82a3841ce8c : SUCCESS in 6m 24s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/f1a8c651cc9d472fa267bd1e5fc3084d : SUCCESS in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb36cc463f4af1f4c3587b82c481d64d3df779c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-14 09:51:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a0455b91115b56305bf6e138ca51a180a9e3c475","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 10:44:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d31b3fb0abb44f9b945c1c6712d6d501 : SUCCESS in 1m 45s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/c74f88d9a45b4b36a7b254f62321451a : SUCCESS in 5m 05s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/29d2b2affa5a4fadbc1dc4f638d449cd : FAILURE in 8m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e64d809c95104197830e0c8b5d4572d0 : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fb58ce2761242e1a46726c5c93443d9 : FAILURE in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/280fb7c6cb32440f8807df426cb9ea1a : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b25ded3e4ba4b2b8ba6c833c25efca3 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c273486c7eb04f549c60ff5c4057b84e : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/710f233b720441788a8aa1fb2f53ef6b : SUCCESS in 6m 28s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/ffc82e0848944122b21c9f7addbb6f55 : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57e54258adc15a5a02b56e9168fede3757e6f76f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-14 10:51:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"970933a6718dad7535534470581913147ecf829b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 12:03:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d9a4f4ab197c4edda51df2d93643c8f9 : SUCCESS in 2m 19s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/2aa21e8cf5a749d7b93f2924e8a6ccdf : SUCCESS in 5m 21s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/0156805732184666aeac22d356339b32 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/475a41caa5bb480e9c61bccfe798c6ed : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b33f847b633d44fd903f3c67e05c1348 : FAILURE in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6cc33b413484dd5a84fe2f5d5272a2e : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e0a546e6e4943a09b36145721ce4afc : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5a4bc0de45b46809f8158e72e9e86f5 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/772d9eb4b7fa4b4182699282a6df58d1 : SUCCESS in 6m 21s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/7c77695d21454ec08f21d9bc29590f2d : SUCCESS in 54m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f5624d198df3915bf5833e61d68553650526d6ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-14 14:43:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f67ab76d4461adcfab6bf0e47c998d2b8bae5666","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-14 14:44:46.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6313675f73df4b3429b59b03863c1c5263b54f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 15:24:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/da1ddd9127ac4a619e63f5b4d42f2477 : SUCCESS in 2m 20s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/06c66030b8d440d8acdfd70d2f99dc32 : SUCCESS in 5m 14s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/7dd239a9a7854177a7ac72281d0af235 : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5a6e5068b334446bb47d408964c29be : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a3ba48ecc48453dbee83f0ff0cc2f90 : FAILURE in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8e0afb8b1c5429ab7bfca3d4223edf3 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56b4471117854c669414675652aa85cd : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5a6a1c4c3954bd4a2478975993e4102 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99bfa23a0c3b4fdaad866067efdc4d59 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74e40b8c06894274ad147d5dca68ba78 : SUCCESS in 5m 06s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/a2f7051e365c494c96035934cbcb06e5 : SUCCESS in 37m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d6a532b1e1f9fc2600c60c5e47ccdf909a7b9ade","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-16 05:31:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"105cbf11c6f78c3eb2673b2a7d7ce7bff4f6379d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-16 05:34:34.000000000","message":"Topic migrate-to-focal removed","accounts_in_message":[],"_revision_number":5},{"id":"320d23395959f42d48c4b66e7b8ac67776c9ad0e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-16 05:35:00.000000000","message":"Topic set to remove-chrome","accounts_in_message":[],"_revision_number":5},{"id":"510b2f8dde82cb6b2b87bcc6960f69c4c8ceec74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 06:52:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/48167469933d4e1a9a61298712a92716 : SUCCESS in 2m 23s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/475440adedea4ada91286c28ef3d0fde : SUCCESS in 6m 24s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/5d9e11c24f9146fd84d3bd9224e37b40 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a86a038e6bb14bca833e8328dd7f1585 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/416c3a65209648df9013713ad48acdb7 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af0154f3846447e881887cb7d99116de : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66c28ea65eae4624abe2cee543fec083 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27fa6208c685425e9fbb87c830bb3a75 : SUCCESS in 7m 24s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/1c1d3ab7399b47a6a373b9ce19da2ae2 : SUCCESS in 1h 18m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e5599a389607ad1356bdfc51aca4184c4a3cfcfb","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-16 10:18:06.000000000","message":"Patch Set 5: Code-Review+2\n\nthanks!","accounts_in_message":[],"_revision_number":5},{"id":"2c6562e91c6eaa2c3270d8cb6334e7b4a5d6e3b7","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-16 16:38:45.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5: Code-Review+2\n\u003e \n\u003e thanks!\n\nyw btw what about stable/victoria because I think lower-constraints also\nfails in stable branches as well. Ironic team also planned to drop l-c job\nfrom stable branches as well?","accounts_in_message":[],"_revision_number":5},{"id":"3417c1d39b3b868da58a0605f691038e2de01feb","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2020-12-16 21:51:51.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1989ae7387294c54fda9e04330f9026b0c4b75fb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-18 08:46:02.000000000","message":"Topic remove-chrome removed","accounts_in_message":[],"_revision_number":5},{"id":"61e2e6547919d583beaf1e73bebff8a173cf6c96","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-18 08:46:10.000000000","message":"Topic set to remove-l-c-job","accounts_in_message":[],"_revision_number":5},{"id":"a93a31c6bbfa1d48f1e3e6fbcfa9c135c33275b1","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-18 08:47:48.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e \u003e Patch Set 5: Code-Review+2\n\u003e \u003e \n\u003e \u003e thanks!\n\u003e \n\u003e yw btw what about stable/victoria because I think lower-constraints also\n\u003e fails in stable branches as well. Ironic team also planned to drop l-c job\n\u003e from stable branches as well?\n\nwhen this merges, feel free to remove the lower-constraints job from stable branches up to train, use the topic remove-l-c-job to ease tracking\nplease DO NOT remove the lower-constraints.txt file in the stable branches\nthanks!","accounts_in_message":[],"_revision_number":5},{"id":"6195f5c7baa019026c3297586711123bfb5ace42","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2020-12-18 13:32:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThanks, Vishal!","accounts_in_message":[],"_revision_number":5},{"id":"7d674949529894033d47bdc39ffc655a2ad6b4db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 13:33:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"aa67050df287a70991cae14037e1ca011fdd48ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 13:54:59.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9cfb5cf6e8084ee3ada9dfdf57b3fc94 : SUCCESS in 2m 54s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/c24fffb3a2564dfda8d136b27cd243ae : SUCCESS in 7m 24s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/05b9c91abcb743f09b5ebd7aaf2a0211 : SUCCESS in 13m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04cbf3c693f240578e6cb05c4aec4f5a : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d15ab9a5933d42abb8e5ff8c4725b882 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/557117133c834db882af36404072df94 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36cc83289f9f4775a594bdc0c3b50f75 : SUCCESS in 15m 35s","accounts_in_message":[],"_revision_number":5},{"id":"bfbc03edde61a3e1de16f4e54f7999413a2de4ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 13:55:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"4491fd420cacb468c30d78e4b978aa103ce5817d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:00:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0dea9ae62d84bcb9fa448970004c7bb : SUCCESS in 55s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a3d0f4e62672cba5dde6e211bb9a44964be164af","revisions":{"b39bff5e3989fc0c2c16ea80e0a9b0bf26be8ab1":{"kind":"REWORK","_number":1,"created":"2020-12-14 05:26:39.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/80/766880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/80/766880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/1"}}},"commit":{"parents":[{"commit":"a11375c2c8cf5d4f741290058376e9e7db2f3348","subject":"Merge \"Remove hard-coded Python path in karma.conf.js\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a11375c2c8cf5d4f741290058376e9e7db2f3348"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 05:25:10.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 05:25:10.000000000","tz":0},"subject":"[DNR] Test nodejs-job","message":"[DNR] Test nodejs-job\n\nChange-Id: Ic7094ab6ed01af2bcbe0beb13f300499215fd903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/b39bff5e3989fc0c2c16ea80e0a9b0bf26be8ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/b39bff5e3989fc0c2c16ea80e0a9b0bf26be8ab1"}]},"branch":"refs/heads/master"},"520aa11204d3cc8cf283627662150fc261ec687b":{"kind":"REWORK","_number":2,"created":"2020-12-14 09:51:06.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/80/766880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/80/766880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/2"}}},"commit":{"parents":[{"commit":"a11375c2c8cf5d4f741290058376e9e7db2f3348","subject":"Merge \"Remove hard-coded Python path in karma.conf.js\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a11375c2c8cf5d4f741290058376e9e7db2f3348"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 05:25:10.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 09:50:50.000000000","tz":0},"subject":"[DNR] Test nodejs-job","message":"[DNR] Test nodejs-job\n\nChange-Id: Ic7094ab6ed01af2bcbe0beb13f300499215fd903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/520aa11204d3cc8cf283627662150fc261ec687b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/520aa11204d3cc8cf283627662150fc261ec687b"}]},"branch":"refs/heads/master"},"a5137df9c0e8d0eac3561d12f8e846346755bf21":{"kind":"REWORK","_number":3,"created":"2020-12-14 10:51:50.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/80/766880/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/80/766880/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/3"}}},"commit":{"parents":[{"commit":"a11375c2c8cf5d4f741290058376e9e7db2f3348","subject":"Merge \"Remove hard-coded Python path in karma.conf.js\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a11375c2c8cf5d4f741290058376e9e7db2f3348"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 05:25:10.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 10:51:36.000000000","tz":0},"subject":"[DNR] Test nodejs-job","message":"[DNR] Test nodejs-job\n\nChange-Id: Ic7094ab6ed01af2bcbe0beb13f300499215fd903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a5137df9c0e8d0eac3561d12f8e846346755bf21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a5137df9c0e8d0eac3561d12f8e846346755bf21"}]},"branch":"refs/heads/master"},"b9c7a3600ccc962162701f53f2ac39dfca7cd3e3":{"kind":"REWORK","_number":4,"created":"2020-12-14 14:43:44.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/80/766880/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/80/766880/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/4"}}},"commit":{"parents":[{"commit":"a11375c2c8cf5d4f741290058376e9e7db2f3348","subject":"Merge \"Remove hard-coded Python path in karma.conf.js\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a11375c2c8cf5d4f741290058376e9e7db2f3348"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 05:25:10.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 14:42:16.000000000","tz":0},"subject":"Remove Chrome browser from karma.conf.js","message":"Remove Chrome browser from karma.conf.js\n\nThis patch removes Chrome browser from\nkarma.conf.js as nodejs job fails with Chrome\nbrowser because it is not able to access some URL.\nSo let\u0027s remove Chrome browser now then we can add\nit again once the issue is resolved.\n\nIt also bump hacking version to 4.0.0 to fix\nlower-constraints job.\n\nChange-Id: Ic7094ab6ed01af2bcbe0beb13f300499215fd903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/b9c7a3600ccc962162701f53f2ac39dfca7cd3e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/b9c7a3600ccc962162701f53f2ac39dfca7cd3e3"}]},"branch":"refs/heads/master"},"a3d0f4e62672cba5dde6e211bb9a44964be164af":{"kind":"REWORK","_number":5,"created":"2020-12-16 05:31:51.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/80/766880/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/80/766880/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/80/766880/5"}}},"commit":{"parents":[{"commit":"a11375c2c8cf5d4f741290058376e9e7db2f3348","subject":"Merge \"Remove hard-coded Python path in karma.conf.js\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a11375c2c8cf5d4f741290058376e9e7db2f3348"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 05:25:10.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-16 05:29:25.000000000","tz":0},"subject":"Remove Chrome browser from karma.conf.js","message":"Remove Chrome browser from karma.conf.js\n\nThis patch removes Chrome browser from\nkarma.conf.js as nodejs job fails with Chrome\nbrowser because it is not able to access some URL.\nSo let\u0027s remove Chrome browser now then we can add\nit again once the issue is resolved.\n\nAlso as discussed during the upstream ironic community meeting on\nMonday Dec 14 2020, the lower-constraints job is being removed.\n\nCo-Authored-By: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nChange-Id: Ic7094ab6ed01af2bcbe0beb13f300499215fd903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a3d0f4e62672cba5dde6e211bb9a44964be164af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a3d0f4e62672cba5dde6e211bb9a44964be164af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
