)]}'
{"id":"openstack%2Fironic-ui~767097","triplet_id":"openstack%2Fironic-ui~master~I8193bdacf2c40f12e133c8ef17baa80b7375cebf","project":"openstack/ironic-ui","branch":"master","topic":"remove-l-c-job","hashtags":[],"change_id":"I8193bdacf2c40f12e133c8ef17baa80b7375cebf","subject":"Remove lower-constraints job","status":"ABANDONED","created":"2020-12-15 08:29:50.000000000","updated":"2020-12-16 10:19:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2a5bdbca384f1d0aaadea05b296b65c68cbc82b5","_number":767097,"virtual_id_number":767097,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-15 18:37:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-12-15 18:04:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 09:22:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-15 18:04:37.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-12-16 05:25:17.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"CC"}],"messages":[{"id":"1e827ae542f835c08786523895a54ce8c0156dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-15 08:29:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d9ea1004b987fe82f5c4e92b75cd156dd7643fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 09:22:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/8cf07ad859214eadb6af41f384a65b0c : SUCCESS in 5m 28s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/521f04a6707243708141f1025891f92b : FAILURE in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a67e88a574044d6b850d43311804e07c : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7026361063f4ffd9c8a65ec90ebe6e9 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/155a821eea8f4378b6f2d58c8302698e : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/856d18a89ea8470bbcf1d46da3b383a2 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0edd45fad9a47549be1b3bf26eebed3 : SUCCESS in 10m 06s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/d4cb28e4761449bebd6a42f57c4c4e72 : SUCCESS in 45m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a4ea358205eb543b824a6bdbb59a476e5333cdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-15 17:37:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"497b886ab786ae217c3391ccc9dc81a7bb1f0296","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-15 18:04:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f37d61680487284184fb5e0c27e84e3194e2d965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 18:37:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3ba0391527e04ba8887f9b976e9bba4d : SUCCESS in 2m 32s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/25786b6446f34e33b4efc7753dbd8e45 : SUCCESS in 5m 18s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/66cc5faaa07a4e7c8d1d262bd3caa4e2 : FAILURE in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/002f3a38f4eb4ae5a7f14a557f8ae231 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88bc1d522feb4bbaad4f970d10272909 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c07b7d5c02494e51a326d0f16c9f0c00 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bf563ce37674cc9b93c067a83215483 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca8f4a34e754414ab86f8fe765a21cb1 : SUCCESS in 6m 39s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/08f3cc5c992a4880beb5633cf5b0f6a7 : SUCCESS in 42m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b6c38830ecb0b84b13dce5299616c5aad3fb9e1","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-16 05:25:17.000000000","message":"Patch Set 2:\n\nlooks like we have to squash this patch with https://review.opendev.org/c/openstack/ironic-ui/+/766880","accounts_in_message":[],"_revision_number":2},{"id":"818537deaa3ca490d5f5837cea75dfd06a9ccdd2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-16 05:33:57.000000000","message":"Patch Set 2:\n\nI have squashed this patch with mine https://review.opendev.org/c/openstack/ironic-ui/+/766880\nto unblock the gate. please take a look.","accounts_in_message":[],"_revision_number":2},{"id":"d765a8c69af451ef383daed4e0f415962037e222","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-16 10:18:42.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e I have squashed this patch with mine https://review.opendev.org/c/openstack/ironic-ui/+/766880\n\u003e to unblock the gate. please take a look.\n\nthanks, I will abandon this","accounts_in_message":[],"_revision_number":2},{"id":"2a5bdbca384f1d0aaadea05b296b65c68cbc82b5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-16 10:19:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1dff8129538b1c1ebcd2247c47ea6ab000f5578b","revisions":{"ed1bf1070c6856a9a62afe97506e642bb9616a3a":{"kind":"REWORK","_number":1,"created":"2020-12-15 08:29:50.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/767097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/97/767097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/97/767097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/97/767097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/97/767097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/97/767097/1"}}},"commit":{"parents":[{"commit":"a11375c2c8cf5d4f741290058376e9e7db2f3348","subject":"Merge \"Remove hard-coded Python path in karma.conf.js\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a11375c2c8cf5d4f741290058376e9e7db2f3348"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-15 08:29:36.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-15 08:29:36.000000000","tz":60},"subject":"Remove lower-constraints job","message":"Remove lower-constraints job\n\nAs discussed during the upstream ironic community meeting on\nMonday Dec 14 2020, the lower-constraints job is being removed.\nWe keep the lower-constraints file as list of suggested min working\nversions.\n\nChange-Id: I8193bdacf2c40f12e133c8ef17baa80b7375cebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/ed1bf1070c6856a9a62afe97506e642bb9616a3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/ed1bf1070c6856a9a62afe97506e642bb9616a3a"}]},"branch":"refs/heads/master"},"1dff8129538b1c1ebcd2247c47ea6ab000f5578b":{"kind":"REWORK","_number":2,"created":"2020-12-15 17:37:20.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/767097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/97/767097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/97/767097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/97/767097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/97/767097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/97/767097/2"}}},"commit":{"parents":[{"commit":"a11375c2c8cf5d4f741290058376e9e7db2f3348","subject":"Merge \"Remove hard-coded Python path in karma.conf.js\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a11375c2c8cf5d4f741290058376e9e7db2f3348"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-15 08:29:36.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-15 17:37:12.000000000","tz":60},"subject":"Remove lower-constraints job","message":"Remove lower-constraints job\n\nAs discussed during the upstream ironic community meeting on\nMonday Dec 14 2020, the lower-constraints job is being removed.\n\nChange-Id: I8193bdacf2c40f12e133c8ef17baa80b7375cebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/1dff8129538b1c1ebcd2247c47ea6ab000f5578b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/1dff8129538b1c1ebcd2247c47ea6ab000f5578b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
