)]}'
{"id":"openstack%2Fironic-ui~785877","triplet_id":"openstack%2Fironic-ui~master~I3616e1fb4d6c7fea29c8ab9d5383ebda2b1b08af","project":"openstack/ironic-ui","branch":"master","topic":"migrate-to-nodejs14","hashtags":[],"change_id":"I3616e1fb4d6c7fea29c8ab9d5383ebda2b1b08af","subject":"Add horizon-nodejs14-job template.","status":"MERGED","created":"2021-04-12 13:34:02.000000000","updated":"2021-05-03 13:59:55.000000000","submitted":"2021-05-03 13:57:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785877-migrate-to-nodejs14","meta_rev_id":"1f08be2d260cefe5c65dbf904481437174312203","_number":785877,"virtual_id_number":785877,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-03 13:57:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-03 13:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"date":"2021-05-03 11:38:03.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-03 13:47:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-12 14:12:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 06:19:36.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-04-29 06:19:36.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-04-29 06:19:36.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-05-03 13:47:19.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"b198cb88624cf7fdca7c22ca1f25d146f7f56bf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-04-12 13:34:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb868cab219b4f02e795ef2a46317c28b80bc868","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-04-12 13:34:19.000000000","message":"Topic set to migrate-to-nodejs14","accounts_in_message":[],"_revision_number":1},{"id":"db85ec2a82b13584b26e082917189079df107e75","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-04-12 13:34:36.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f912614a30e09c935a85d1e4d6ded2176cca203e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 14:12:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f9a9812166346ef9da67e88fdb14cb2 : SUCCESS in 4m 28s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/085d51d1167f4e8285df2e8de0d3a1cd : SUCCESS in 6m 03s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/8ab6ef2e129542b2b5e088786ab0184e : SUCCESS in 8m 56s\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/cdbda9d39ea54da19e42265c2ac2567f : SUCCESS in 6m 49s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/ce7fd265f7e04e7a91afbd0a08040897 : SUCCESS in 12m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c73949f0e6a4045b2f58aba83bb46b0 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f86d84c3f3674f7495a0dd5548935238 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80a2979af4ac48b29b03dce978b89d86 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/301d22f51b12471192e1aeb6b2ee62e9 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a24ddad75a40493dbc39b2da6bcbb457 : SUCCESS in 8m 46s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/c49efeaa42f945ef9d9cdac64df933bd : SUCCESS in 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed118fa4c9c9abaa5637565eaa904781336bd501","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-04-29 06:19:40.000000000","message":"Removed Workflow-1 by Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"b826dd32caf3f5321be5d5f3dad8d02d12be0782","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-05-03 11:38:03.000000000","message":"Patch Set 1:\n\nHi all, Could you review this patch because horizon team is going to drop\nthe nodejs10 job once this patch is merged as nodejs10 is already EOL.\nthanks.","accounts_in_message":[],"_revision_number":1},{"id":"9829204cd9882047dfa1ba1dd93c580047197cf5","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-05-03 13:47:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivial CI change, approving","accounts_in_message":[],"_revision_number":1},{"id":"1bba8afc447a3d7ac9f838098ded748712acb151","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 13:47:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f918838c9ccb5206b74e22d73ac8cf7778d0889b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 13:57:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/01541665b494473f9fc33659731f6581 : SUCCESS in 6m 18s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/5f6f33afe4464e22bb39de9ea488c0d0 : SUCCESS in 8m 04s\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/a37a32ddc40b472da5b3f912e5db1f09 : SUCCESS in 5m 12s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/16bc814452834ed08a885defdc72590b : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b13c68cee7c44edac1f87f6f8fad6c3 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2402865951441e3aee6dc6f4e7552cc : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4d3c69e56a94dbf9805cd55cbeb1618 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2dbb4981347496487cd33ac6a31f7a9 : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0d63774a6573017e4565363de66459902d58d61d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 13:57:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1f08be2d260cefe5c65dbf904481437174312203","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 13:59:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30e3d26931043f497f61ec7c4ea4232 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0e3256a20b253e07521ec33806222a27746af2f","revisions":{"c0e3256a20b253e07521ec33806222a27746af2f":{"kind":"REWORK","_number":1,"created":"2021-04-12 13:34:02.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/77/785877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/77/785877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/77/785877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/77/785877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/77/785877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/77/785877/1"}}},"commit":{"parents":[{"commit":"f219d0b1cbbf00a5b3d05d593288bed35157f9e3","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/f219d0b1cbbf00a5b3d05d593288bed35157f9e3"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-04-12 13:31:09.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-04-12 13:33:59.000000000","tz":330},"subject":"Add horizon-nodejs14-job template.","message":"Add horizon-nodejs14-job template.\n\nDepends-On: https://review.opendev.org/c/openstack/horizon/+/784062\nChange-Id: I3616e1fb4d6c7fea29c8ab9d5383ebda2b1b08af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/c0e3256a20b253e07521ec33806222a27746af2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/c0e3256a20b253e07521ec33806222a27746af2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
