)]}'
{"id":"openstack%2Fironic-ui~839971","triplet_id":"openstack%2Fironic-ui~master~I4d64f8ecd2fd665dd4133603dce078e043d00177","project":"openstack/ironic-ui","branch":"master","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2022-05-04 09:48:36.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-05-04 09:37:13.000000000","reason":"Riccardo Pittau replied on the change"},"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2022-05-04 09:48:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d64f8ecd2fd665dd4133603dce078e043d00177","subject":"Address RemovedInDjango40Warning","status":"MERGED","created":"2022-04-29 16:31:02.000000000","updated":"2022-05-04 09:49:55.000000000","submitted":"2022-05-04 09:48:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839971","meta_rev_id":"62e3b2c1bcfa41eee181f2552311e6836ac19be5","_number":839971,"virtual_id_number":839971,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-04 09:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-05-04 09:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2022-05-04 09:37:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 17:23:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-02 10:55:51.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-05-02 10:55:51.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"d2ab99b40921580a4d347a9a0c7da4cce5cd49f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-04-29 16:31:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6abb92050c47824dfc8e557900e18165b081ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 17:23:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcb8c9d8d7ed4ec9b7cea8f605fa015d : SUCCESS in 4m 59s\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/15cdf35e9e51445688f2b8b4747adc48 : SUCCESS in 5m 32s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/0db49b25e8334c7ba15778cb44b7f51c : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d84431e981b47729b37be787b13dc47 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e17385b9cb3f4381bd2ca56653099d82 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/583dc8b6dc1b4f8e86fa0f6504908cf5 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd462fae9efa476c910a341da92770d5 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb024f182f3e4be79911fff87d65c8de : SUCCESS in 7m 36s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/a98651d8ae434720b8eb0381f6a41dd7 : FAILURE in 45m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60618f44d69f510f70b8b2d9195ac61935e0dfe4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-05-04 09:37:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef22dbdfa60f20904a9796e8f40699905c0e8f53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 09:37:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb5d09bf000eb3cf1006639ee4c57a712991fa1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 09:48:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/fa4c60b05f0d44ad934528dd54c77d4d : SUCCESS in 4m 54s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/2296acd5dc364b86b498c5483b46c2c7 : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7b1d99d79d94f0596a0bd1ee4e8e5ec : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41100e053bf42f0bbb32833d4ed5482 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e65fac785ab4380a3a3311d6470aece : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/410c87381fbb4f8db56aea2dc51c89a9 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7f9a525692497ab782192e645066dbf115807622","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 09:48:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"62e3b2c1bcfa41eee181f2552311e6836ac19be5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 09:49:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c959a9ab670049118e989fe8e2ad36bd : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16c4d0a63c22c5f7189967cfb55341811bcdc44e","revisions":{"16c4d0a63c22c5f7189967cfb55341811bcdc44e":{"kind":"REWORK","_number":1,"created":"2022-04-29 16:31:02.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/71/839971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/71/839971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/71/839971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/71/839971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/71/839971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/71/839971/1"}}},"commit":{"parents":[{"commit":"74a0b9c4953e4319ce25badcd7338985f7e1933c","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/74a0b9c4953e4319ce25badcd7338985f7e1933c"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-04-29 16:25:18.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-04-29 16:30:59.000000000","tz":330},"subject":"Address RemovedInDjango40Warning","message":"Address RemovedInDjango40Warning\n\nIn Django 3.1, django.conf.urls.url() is deprecated\nin favor of django.urls.re_path(). For more info see [1]\nThese were already replaced in Horizon repo by [2].\n\n[1] https://docs.djangoproject.com/en/4.0/releases/3.1/#id2\n[2] https://review.opendev.org/c/openstack/horizon/+/827093\n\nChange-Id: I4d64f8ecd2fd665dd4133603dce078e043d00177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/16c4d0a63c22c5f7189967cfb55341811bcdc44e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/16c4d0a63c22c5f7189967cfb55341811bcdc44e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
