)]}'
{"id":"openstack%2Fironic-ui~846849","triplet_id":"openstack%2Fironic-ui~master~Ia14430c8d82342a704a24394dcaa90968aa117ee","project":"openstack/ironic-ui","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ia14430c8d82342a704a24394dcaa90968aa117ee","subject":"Imported Translations from Zanata","status":"MERGED","created":"2022-06-21 02:58:11.000000000","updated":"2022-08-19 12:50:35.000000000","submitted":"2022-08-19 12:48:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846849-zanata/translations","meta_rev_id":"91ffc47e9feb82cb4428e712222a3168c78c4e06","_number":846849,"virtual_id_number":846849,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-19 12:48:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-19 12:13:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-19 12:13:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-21 03:18:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-19 12:13:55.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"9f4fc1c8db8f187370f04c0b4d80f2f3be13840d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-21 02:58:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f9d03ae064b1534dc031ad5209591d96e5a4b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 03:18:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/6566beeba6da4fef89d7b4c88b0fa290 : SUCCESS in 4m 59s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/81628d500eb64065bf61b3519bba17e7 : FAILURE in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aaab32ffcf84dbca1b2aad22e5e641f : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/111faef81b9543819f6cfa990b8d3f9b : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4eda80a3c4664eceae4b09001dbdb112 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0d2b94132b6c258de90c0ba8856e701ed838d323","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-19 12:13:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"79655f7dcd1d67d7c1ef9220bac5247ffecd3a1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 12:26:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b5b93e5586478a95a353e42cfd3ba1\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/028c40242bd347ae9a8fce3adfdcbc1f : SUCCESS in 6m 05s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/a94963c2ad504aae9fe65afdc55474c9 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90cdbb2509a4413b897db16bd137c88f : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d579979ea594b56982b89251e05556d : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fef98417977b49609c18e2ff171d5c9f : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cef573edc4340b9670dcd7551c2ee0bdc30548e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 12:26:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cbb3fff5d178cdbd1135cac2abe652c9d0fae1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 12:48:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6514597c79914468b5507e3b4254110a\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/d92a593a3e464385b0152a9bc13a1352 : SUCCESS in 4m 28s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/4614738428cb4eb0a2ecef14a87646e6 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7eb6f4af0b14efb95845cfa69a3beb5 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfcdc21a857947b38c8cad641c386f12 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bb9d51af7854cc281cf93b1ecccf413 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5912680133608c086e5ff7cfceaedba8ccac422a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 12:48:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"91ffc47e9feb82cb4428e712222a3168c78c4e06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 12:50:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbb2c0d343f8401084d0672f98a976bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d006bf02b04971aa3e5a0c41dd64cf : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ca977bf1ab74489a879490add958058 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e49dc3525cde0634072818e1e40f8b0364c6ac1","revisions":{"5e49dc3525cde0634072818e1e40f8b0364c6ac1":{"kind":"REWORK","_number":1,"created":"2022-06-21 02:58:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/846849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/49/846849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/49/846849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/49/846849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/49/846849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/49/846849/1"}}},"commit":{"parents":[{"commit":"8974cb6771a663d5be146fabaec2abd5982426e8","subject":"Merge \"Use Horizon project template for nodejs jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/8974cb6771a663d5be146fabaec2abd5982426e8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-21 02:58:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-21 02:58:09.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ia14430c8d82342a704a24394dcaa90968aa117ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/5e49dc3525cde0634072818e1e40f8b0364c6ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/5e49dc3525cde0634072818e1e40f8b0364c6ac1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[]}
