)]}'
{"id":"openstack%2Fironic-ui~852314","triplet_id":"openstack%2Fironic-ui~master~Ic652284d841ad7ec012212abc9b570e9fdfd4c1c","project":"openstack/ironic-ui","branch":"master","attention_set":{},"removed_from_attention_set":{"35067":{"account":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"last_update":"2022-08-19 06:22:26.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic652284d841ad7ec012212abc9b570e9fdfd4c1c","subject":"remove unicode from code","status":"ABANDONED","created":"2022-08-06 01:00:14.000000000","updated":"2022-08-19 06:22:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0c849b510b81a2ef6df1ac7242c3e57b2f00f2bb","_number":852314,"virtual_id_number":852314,"owner":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-06 01:10:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-06 01:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8362abefb78f554d13cfe996a773687a8d9f187","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-06 01:00:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7104dc85a24457b605e0b8fa0c52cd1298b8cb94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 01:10:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6354acd305364dc6934dd62d9a76efa5\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/63e26a6f563f4536acd12b61dd3c221a : SUCCESS in 4m 49s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/021fe1ea492c47f594afa60273685911 : FAILURE in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e6dae3528aa41f48f95a2c9350cac6b : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea34598ae19940a8970dcecb7d11b93e : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0c849b510b81a2ef6df1ac7242c3e57b2f00f2bb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-19 06:22:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e618b2bac3601808753cee7694322cda3ee9161d","revisions":{"e618b2bac3601808753cee7694322cda3ee9161d":{"kind":"REWORK","_number":1,"created":"2022-08-06 01:00:14.000000000","uploader":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"ref":"refs/changes/14/852314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/14/852314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/14/852314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/14/852314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/14/852314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/14/852314/1"}}},"commit":{"parents":[{"commit":"8974cb6771a663d5be146fabaec2abd5982426e8","subject":"Merge \"Use Horizon project template for nodejs jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/8974cb6771a663d5be146fabaec2abd5982426e8"}]}],"author":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-06 00:59:53.000000000","tz":480},"committer":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-06 01:00:01.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: Ic652284d841ad7ec012212abc9b570e9fdfd4c1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/e618b2bac3601808753cee7694322cda3ee9161d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/e618b2bac3601808753cee7694322cda3ee9161d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
