)]}'
{"id":"openstack%2Fironic-ui~852618","triplet_id":"openstack%2Fironic-ui~master~Iabb7fc95cc343ea3ef1a81d780eef21881bf6bbe","project":"openstack/ironic-ui","branch":"master","attention_set":{},"removed_from_attention_set":{"35067":{"account":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"last_update":"2022-08-19 06:22:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iabb7fc95cc343ea3ef1a81d780eef21881bf6bbe","subject":"remove unicode from code","status":"ABANDONED","created":"2022-08-11 06:48:26.000000000","updated":"2022-08-19 06:22:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d422237fec854cadd310612eb0270a78e6711bff","_number":852618,"virtual_id_number":852618,"owner":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-11 06:57:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-11 06:57:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a5a8e1fdd73bb74647fcf83a8485dafe2e3ba37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-11 06:48:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42a63338e3dd4bae15b4026fa97c8b42fc4adcfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 06:57:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6d52c487cfe49309da667dec66af0ea\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/70b581062cd64292b151c64da6b1fd70 : SUCCESS in 5m 56s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/d6f9a0e3a203497d8f766abf2ddb3bf5 : FAILURE in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9747ddc476049798111523ec6e5cc98 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72238ad108b34ecaaa19c009e2cfb6ff : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d21acdbebe9d469e9986640e3162be94 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d422237fec854cadd310612eb0270a78e6711bff","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-19 06:22:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4376a2e2eea270417723dcbe91da2c2455da060","revisions":{"a4376a2e2eea270417723dcbe91da2c2455da060":{"kind":"REWORK","_number":1,"created":"2022-08-11 06:48:26.000000000","uploader":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"ref":"refs/changes/18/852618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/18/852618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/18/852618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/18/852618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/18/852618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/18/852618/1"}}},"commit":{"parents":[{"commit":"e618b2bac3601808753cee7694322cda3ee9161d","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/e618b2bac3601808753cee7694322cda3ee9161d"}]}],"author":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-11 06:48:06.000000000","tz":480},"committer":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-11 06:48:13.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: Iabb7fc95cc343ea3ef1a81d780eef21881bf6bbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a4376a2e2eea270417723dcbe91da2c2455da060"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a4376a2e2eea270417723dcbe91da2c2455da060"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
